)]}'
{"id":"openstack%2Felection~855527","triplet_id":"openstack%2Felection~master~I49e5900d6b004932f7108376054c6847a5e6706b","project":"openstack/election","branch":"master","topic":"a_rally","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2022-09-02 11:04:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I49e5900d6b004932f7108376054c6847a5e6706b","subject":"Adding andreykurilin candidacy for Rally PTL","status":"MERGED","created":"2022-09-01 15:25:06.000000000","updated":"2022-09-02 11:05:55.000000000","submitted":"2022-09-02 11:04:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855527-a_rally","meta_rev_id":"a27b6c265cb40856b26c40783e8a59578a28ee8a","_number":855527,"virtual_id_number":855527,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-02 11:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2022-09-02 10:55:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2022-09-01 21:36:43.000000000","_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2022-09-02 10:55:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-01 15:34:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-01 21:36:43.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2022-09-02 10:55:47.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"d181d658d831c8f974a7339aa238068927af117b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-09-01 15:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"448291e88afd9fee4c9a9942f8ff1376f2cd4143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 15:34:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aff8bd8709644a1989fd2792ad88faf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60fdb01f7ec14af88e30358fc53bd97b : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93ef2793a2eb468dab8fa3f03a836ab2 : SUCCESS in 3m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/24126d58f8034c66ade0a873ebae2e1e : SUCCESS in 2m 51s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/ab4d80abfd584732b7971d5603cfcd78 : SUCCESS in 4m 16s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b23fbe80fe654c5591e9bf79f569980d : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"11a0345cbe6c396b4b5fd2f19bb03ee44e6871df","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-09-01 21:36:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"253075474c9eb68601d43e542f0aac5ce6d68cc1","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-09-02 10:55:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32af6a2a66e2d00b9fc14fdabd23ba63eb147d1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 10:56:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5e121179a9efefba410e173231e9ebc1af12291","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:04:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfc10bd747354aba89adf98acca2b662\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef7cb5ebc0d4142a8f16fcf765ef88c : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24a972b0ad8d48dab0e12ec1cb426bb3 : SUCCESS in 4m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0108fa92386d47bf8add0e8c949d1870 : SUCCESS in 3m 58s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/fd2ca41438c5472490f32fe8c7da0e4e : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"13ab9055f50929843e2bf21fb449e39516300f80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:04:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a27b6c265cb40856b26c40783e8a59578a28ee8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:05:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ece4691a40b48d98e3983a5c8121eda\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/0056ce19bf994584af5b32b8b4b0f6b6 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e862c9a7564a1435d1edd89016e048675b7aafd","revisions":{"5e862c9a7564a1435d1edd89016e048675b7aafd":{"kind":"REWORK","_number":1,"created":"2022-09-01 15:25:06.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/27/855527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/27/855527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/27/855527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/27/855527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/27/855527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/27/855527/1"}}},"commit":{"parents":[{"commit":"e6554d1cdfa2ca3c2b14f300fd9cef2a471fec37","subject":"Merge \"Add Martin Kopec candidacy for QA Antelope PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e6554d1cdfa2ca3c2b14f300fd9cef2a471fec37"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-09-01 15:10:14.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-09-01 15:24:55.000000000","tz":180},"subject":"Adding andreykurilin candidacy for Rally PTL","message":"Adding andreykurilin candidacy for Rally PTL\n\nChange-Id: I49e5900d6b004932f7108376054c6847a5e6706b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5e862c9a7564a1435d1edd89016e048675b7aafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5e862c9a7564a1435d1edd89016e048675b7aafd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[]}
