)]}'
{"id":"openstack%2Felection~855660","triplet_id":"openstack%2Felection~master~Ibaed4eab2368a590e251c88e90187fa805b6f187","project":"openstack/election","branch":"master","topic":"tc/antelope","attention_set":{},"removed_from_attention_set":{"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-09-02 18:38:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibaed4eab2368a590e251c88e90187fa805b6f187","subject":"Adding chandankumar candidacy for TC","status":"MERGED","created":"2022-09-02 13:05:14.000000000","updated":"2022-09-02 18:39:47.000000000","submitted":"2022-09-02 18:38:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855660-tc/antelope","meta_rev_id":"8920ee5131ddf85147993b6d6ea9162d7f0465c3","_number":855660,"virtual_id_number":855660,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-02 18:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":2,"date":"2022-09-02 16:20:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-02 18:29:17.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-02 18:29:17.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-02 13:16:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-02 16:20:21.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2022-09-02 18:29:17.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"b21dc0083b1c8b5acea792ed5e565e0ca7380e93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-02 13:05:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"635b67d4f73fd5808d58ca797407aa521fd862aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 13:16:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/030a322723ab4c038355f635d421eb54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d26ef99fec439b8952ae1483b3570d : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4f7fb37830e44e2b907f55e43e08f0d : SUCCESS in 2m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/50a006c8532e42d982411aec322cfded : SUCCESS in 3m 41s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/995af43c8e39443492bff3443234d84f : SUCCESS in 2m 35s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/46ff6c3e3811419eb620e08ea56fc8f9 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7d0d9fb9225aa418d7b00a5de2c932a6b368552a","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-09-02 16:20:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a9ebfc4edc5680102b5fbd0af2e0c8250cc5452","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-09-02 18:29:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10ab7eb86c6994dd652a10fc57768a7196195f60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 18:29:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6842e7d9c5095dbda07d009abb32871df5c1285","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 18:38:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80369465eba449bdbae31146c428c124\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2357fb8356fb4a42aaee426c9081032e : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a46d9a87430442cf967a4702eddfe7c1 : SUCCESS in 6m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ec3ce675ce4b4d2d961572ccd166ba23 : SUCCESS in 4m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/838c4e7807ef4c598399b27be53027bf : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"dcbd3127d9b235bdbd217d3b8a88fb51bfe06a3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 18:38:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8920ee5131ddf85147993b6d6ea9162d7f0465c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 18:39:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a468635e6a41a79ae4ef96f65b4986\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/7c72b192a86e43a299f3c31c6cbe6fb4 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2b2639e153eeec53e6c4b0dfb162a3a96ed35e8","revisions":{"c2b2639e153eeec53e6c4b0dfb162a3a96ed35e8":{"kind":"REWORK","_number":1,"created":"2022-09-02 13:05:14.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/60/855660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/60/855660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/60/855660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/60/855660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/60/855660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/60/855660/1"}}},"commit":{"parents":[{"commit":"0bd0f408a14bd70438e07174f372befe0d4c6dc4","subject":"Merge \"Adding andreykurilin candidacy for Rally PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0bd0f408a14bd70438e07174f372befe0d4c6dc4"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-02 13:04:06.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-02 13:05:06.000000000","tz":330},"subject":"Adding chandankumar candidacy for TC","message":"Adding chandankumar candidacy for TC\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ibaed4eab2368a590e251c88e90187fa805b6f187\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c2b2639e153eeec53e6c4b0dfb162a3a96ed35e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c2b2639e153eeec53e6c4b0dfb162a3a96ed35e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}}]}],"submit_requirements":[]}
