)]}'
{"id":"openstack%2Felection~856031","triplet_id":"openstack%2Felection~master~I077a1cb889995d1fcc5a3b00a0b1fa3bbbeeb240","project":"openstack/election","branch":"master","topic":"tc-jamespage","attention_set":{},"removed_from_attention_set":{"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2022-09-06 14:13:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I077a1cb889995d1fcc5a3b00a0b1fa3bbbeeb240","subject":"Add jamespage TC candidacy","status":"MERGED","created":"2022-09-06 08:56:09.000000000","updated":"2022-09-06 14:14:55.000000000","submitted":"2022-09-06 14:13:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856031-tc-jamespage","meta_rev_id":"9b9b1bc4d2f5a6fde3fac8a040a5aa0e786a6e74","_number":856031,"virtual_id_number":856031,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-06 14:13:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-09-06 13:52:51.000000000","_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2022-09-06 14:06:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-09-06 14:06:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 09:09:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 13:52:51.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-09-06 14:06:34.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"f2621be744314200b249628f96acb3c358c8b0a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2022-09-06 08:56:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d19cbdf763d7a022ec6d63261709b3cbc9945d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 09:09:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/443b516d18ae459390a231d9840c55de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7531831406944f20b1898e50c216aebb : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6e57567d0f64f17b8f6471206e45c78 : SUCCESS in 7m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7cc65cc48b9a4fd68d293a8f285caa7f : SUCCESS in 5m 32s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e159b03480a74f70bcdae05d293ace7a : SUCCESS in 5m 43s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ae747afdd74245ca9d4a497517952409 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"94663e77c128f97caef9113106d5bec6d8e8138c","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-09-06 13:52:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9afe06d3e5e1796f56998dde23050f6bbe98b22c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-09-06 14:06:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fa03e755b5f71f40ebca424069438219d768ef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 14:06:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26f6d6921974c7c6b0ff3ded1d4ccdd5e9755f04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 14:13:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/677dbfac86484a07a60894b4fe09a616\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4cc1233471f47508eab1be49e2099a2 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/714de02b74964228a9c2818afee6118e : SUCCESS in 5m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ee45e4a4f9224de69260b2c7220c7251 : SUCCESS in 4m 27s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/357a1c74916f47ab83697a07d1817505 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"df2a982532971f056bf4fda03f0d5b72fbe90011","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 14:13:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9b9b1bc4d2f5a6fde3fac8a040a5aa0e786a6e74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 14:14:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7b989935e214bef95a1c257618eabb4\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/defb815f0e7f4b47a60b234642792f1a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77e5932128a44f1c1eb55e5ed28cd2e4b7676fc0","revisions":{"77e5932128a44f1c1eb55e5ed28cd2e4b7676fc0":{"kind":"REWORK","_number":1,"created":"2022-09-06 08:56:09.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/31/856031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/31/856031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/31/856031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/31/856031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/31/856031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/31/856031/1"}}},"commit":{"parents":[{"commit":"c2b2639e153eeec53e6c4b0dfb162a3a96ed35e8","subject":"Adding chandankumar candidacy for TC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c2b2639e153eeec53e6c4b0dfb162a3a96ed35e8"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-09-06 08:51:19.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@ubuntu.com","date":"2022-09-06 08:53:09.000000000","tz":60},"subject":"Add jamespage TC candidacy","message":"Add jamespage TC candidacy\n\nChange-Id: I077a1cb889995d1fcc5a3b00a0b1fa3bbbeeb240\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/77e5932128a44f1c1eb55e5ed28cd2e4b7676fc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/77e5932128a44f1c1eb55e5ed28cd2e4b7676fc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[]}
