)]}'
{"id":"openstack%2Felection~856350","triplet_id":"openstack%2Felection~master~Idedf339fecb5f83e0a3e2383145c1885c8308875","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"15993":{"account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"last_update":"2022-09-08 03:37:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idedf339fecb5f83e0a3e2383145c1885c8308875","subject":"Adding a campaigning period for TC","status":"MERGED","created":"2022-09-08 00:18:00.000000000","updated":"2022-09-08 03:38:48.000000000","submitted":"2022-09-08 03:37:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"856350","meta_rev_id":"35917bc1d2ca4a55542dd9fc0b03d1fd59d64653","_number":856350,"virtual_id_number":856350,"owner":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-08 03:37:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-09-08 03:27:28.000000000","_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2022-09-08 02:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-09-08 03:27:28.000000000","_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-08 00:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-08 00:32:40.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"CC"},{"updated":"2022-09-08 02:40:03.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2022-09-08 03:27:28.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"f8a78bff884c6eb69af9389cca7dee539b24ed53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-09-08 00:18:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7cba46d825d977cf4e4146c2d07d144a00c6fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 00:26:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26dbe2147d4d45c4b71c361ffd2f54a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/108b924408c6470285f11a37cc69aac9 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a5a2ccb9364473fb1f9018ebadee932 : SUCCESS in 3m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f63d7e71e0f0493cad4f47f0a24db130 : SUCCESS in 5m 55s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d46ff6f2ca564741ad95ff54d64677ae : SUCCESS in 3m 18s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8a440b9d289c44a2b115b697c8d64172 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3bd2289d51416bc3dca1f225655540d990a103b2","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-09-08 00:32:40.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4d86c4726a542ddace0116eadbc496e94a56c007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2022-09-08 01:39:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b43f3f1186aff879c1e0c6acaaee4f2907ca801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 01:49:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17e0d4ac99a04a24ae398b5aaaa0948e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1507363fb61c4315ac017ab6259b469a : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c996567ed3647e9a579924eb24c1ebe : SUCCESS in 4m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7cd025783c7c4046a10b69e4b01697ac : SUCCESS in 3m 22s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7eb2b31e1dfb468891280afcc1addff7 : SUCCESS in 3m 45s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2f57db8c1c6242939158ea70dead632f : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4df95b9eaed4d1863adbaeaf7de79d9d69f29761","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-09-08 02:40:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ceeae9cf1297513aefa417b0ef0b43c4083e611f","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-09-08 03:27:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"62ed78a5f7b19b5862d59dbdca4fca5be897dbe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 03:27:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c3420961508e7b8f89036634e7a9f33d0ee2ff3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 03:37:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06bd794a135448359f6c8c1d2db9ecba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2b5899083c45b0a53a576a5c0077b2 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99d5cb41586f40138afe351042530815 : SUCCESS in 2m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/79c3a15d4070439ba46492797a461af2 : SUCCESS in 3m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/94dc33e600444c038ab5821095228ecc : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"eb41c6a622ca43ce05a002468e75c75f444f7004","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 03:37:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"35917bc1d2ca4a55542dd9fc0b03d1fd59d64653","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 03:38:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dc530828d2e47be82254ae6ff70454e\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/b509f8e134e2486f9df1b583f48e6d77 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a032fd9a478387b7dd14d6d2c9b4a52c2d609ac2","revisions":{"ea4651db2c75bde62822f448516512d49161fbbd":{"kind":"REWORK","_number":1,"created":"2022-09-08 00:18:00.000000000","uploader":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"ref":"refs/changes/50/856350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/50/856350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/50/856350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/50/856350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/50/856350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/50/856350/1"}}},"commit":{"parents":[{"commit":"63c4e12a586149edaeeaa696cda42c47fd947e16","subject":"Merge \"Add Jay Faulkner TC Candidacy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/63c4e12a586149edaeeaa696cda42c47fd947e16"}]}],"author":{"name":"Amy Marrich (spotz)","email":"amy@demarco.com","date":"2022-09-08 00:15:05.000000000","tz":-300},"committer":{"name":"Amy Marrich (spotz)","email":"amy@demarco.com","date":"2022-09-08 00:17:56.000000000","tz":-300},"subject":"Adding a campaigning period for TC","message":"Adding a campaigning period for TC\n\nIn order to give folks time to opt-in as well as to assure the CIVS\nsystem is ready we are adding a seperate` TC campaign back in until 9/12\n\nChange-Id: Idedf339fecb5f83e0a3e2383145c1885c8308875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ea4651db2c75bde62822f448516512d49161fbbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ea4651db2c75bde62822f448516512d49161fbbd"}]},"branch":"refs/heads/master"},"a032fd9a478387b7dd14d6d2c9b4a52c2d609ac2":{"kind":"REWORK","_number":2,"created":"2022-09-08 01:39:11.000000000","uploader":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"ref":"refs/changes/50/856350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/50/856350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/50/856350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/50/856350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/50/856350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/50/856350/2"}}},"commit":{"parents":[{"commit":"63c4e12a586149edaeeaa696cda42c47fd947e16","subject":"Merge \"Add Jay Faulkner TC Candidacy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/63c4e12a586149edaeeaa696cda42c47fd947e16"}]}],"author":{"name":"Amy Marrich (spotz)","email":"amy@demarco.com","date":"2022-09-08 00:15:05.000000000","tz":-300},"committer":{"name":"Amy Marrich (spotz)","email":"amy@demarco.com","date":"2022-09-08 01:38:50.000000000","tz":-300},"subject":"Adding a campaigning period for TC","message":"Adding a campaigning period for TC\n\nIn order to give folks time to opt-in as well as to assure the CIVS\nsystem is ready we are adding a seperate` TC campaign back in until 9/12\n\nChange-Id: Idedf339fecb5f83e0a3e2383145c1885c8308875\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a032fd9a478387b7dd14d6d2c9b4a52c2d609ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a032fd9a478387b7dd14d6d2c9b4a52c2d609ac2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[]}
