)]}'
{"id":"openstack%2Felection~862378","triplet_id":"openstack%2Felection~master~If66602b12eb77b17904427149ab81ce7a21d9492","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-02-01 23:00:35.000000000","reason":"Change was submitted"},"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2022-12-06 22:01:50.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}},"hashtags":[],"change_id":"If66602b12eb77b17904427149ab81ce7a21d9492","subject":"Switch election timing to Wednesday","status":"MERGED","created":"2022-10-21 16:31:17.000000000","updated":"2023-02-01 23:01:37.000000000","submitted":"2023-02-01 23:00:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862378","meta_rev_id":"95c5c88b67e5f1184336339e28dbe6fb5cd23963","_number":862378,"virtual_id_number":862378,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-01 23:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-02-01 22:42:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2023-02-01 22:24:25.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-01 22:24:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2023-02-01 22:42:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 16:40:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 00:55:09.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-11-23 16:45:23.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-12-06 20:32:56.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2022-12-06 22:01:50.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"f274985ae4bed14be83e83fed0954273bd353f64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-10-21 16:31:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d34118f3343f618a7070ccc8e5e9ac82c120a7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:40:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/954897d697e548b6afa0dca64248b59b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1c3d2a93d3415c9816e9e9d6856fb4 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd9233dd07d743c4be29e926116182d9 : SUCCESS in 2m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1937cb3c90e44a0a8aa32bfc98af48c6 : SUCCESS in 2m 32s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b7c8cb5c0ad24e36aec862a396a518b7 : SUCCESS in 3m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e15a888e840f4b069a1dd27eec92a4a9 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"09df15e269a9fe615ed999f632b68b5f4d67c10b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-23 00:55:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0ac3e5686b8448ea51f7e2b6c32d03937a988ec3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-23 16:45:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2de75fd0bc4a45306e15ec78c7330aa510d65fcc","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-12-06 20:32:56.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"496ec674ef643c900d184499313fdab4913368e3","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-12-06 21:25:05.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c0e8e3831cb9b9763a4db7e33c2d8b5107bd32de","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-12-06 22:01:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d92cc2980f5fefba3624080c5b723084f3888e30","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-31 22:11:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"018afef6df34a20e912a95b1c3071b797e432ab3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:12:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f8a0707e6f311020a1c781902b44df6f0419997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:25:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a6a0999dd964a6990c26790c54842c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df70ba4842de474d9ccba6d574f7eba8 : FAILURE in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6bc5c386c7d4bc1a6660887752e83ac : SUCCESS in 2m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/42c3d08011a249fb902f8fdc0f0fb880 : SUCCESS in 2m 36s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/aaca8b3c657e4b4f98945fcf8ad34f87 : FAILURE in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6fd080efaa9709a4225f15553c5c2a0358776ee8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-01 11:01:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"75cd60e0b4c958b21a4df0f004fd9682aa7748ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:07:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9049057a1eb4182998786458a9ebccf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5484242dcf46a085adf39495328cb1 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fe23a7520a14e7f8cf742f5b744e3db : SUCCESS in 3m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/75e6f4069b424065b9ee65270e443d4d : SUCCESS in 3m 21s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/982e21c2672a4db1bdcceb47e47f5a35 : SUCCESS in 2m 35s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/fb47a592af13426cbc4890dc7cf8594f : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"7094b3e6a4400d583e1237e7d3271189d7ecb775","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 14:26:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2f48a29ecfab2b47d0f9ae4d4281028d8324de8e","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 21:14:49.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"11d6ada20d29e016dcc47eba1184d6b7a563b099","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 21:14:53.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"42cb0dd87ce33c1cdc16455b4230c09ef4e4fdb2","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 21:19:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43eeb498f751f65a47605fb74857233edc6ee2ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:30:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3efa80f76d154856ac048cc39cc2b0f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abfd938e62e84ebe89206d1846879ad8 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/855eb03d4ef14abd8ed4181f3db6febd : SUCCESS in 2m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/71147ef5d5354823a623496a45131bf8 : SUCCESS in 2m 35s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/bc0c8e076c6c429188bc60a8a0a91328 : SUCCESS in 3m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/50bf76c6e5224b439068b099f6f0f31f : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c5332cb75a01b5e962a630e63ad88c31fd47dc83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 22:24:25.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5beb16c74c50271e685ecfd0ec38d591437beb0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 22:29:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ef97b247a9d4279b9935fb557761b94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc9e5b253ce4825b436a6880bf577f2 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da6f3b2f6b624faf87d5b551de56557b : SUCCESS in 2m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/671be36a90594c9d97cd449f38db0320 : SUCCESS in 3m 20s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/fb2dfb654463475494fb2e200819d528 : SUCCESS in 2m 34s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/16bd1d4a2dc14cebabec882299d6c8ae : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":3},{"id":"51b75408aad376cc5593a52ace48d421c0dc353d","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 22:42:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5efa7bc66bd7e7b4790929c2167291c916607e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 22:42:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94006a4aafbb1fd900c198738a468f5b580a1ab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 23:00:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/889ae81cb1f14f789b0cac02f0509f4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca595d473c6f4eeb94425a95eb3fdc7f : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36fbaea5f5a6440d96463f4b687a1468 : SUCCESS in 3m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9e599de77c50454993c6e6e0ab7ef338 : SUCCESS in 2m 45s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7f44ac8d138f4ca5a9b835a9e71fe265 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b12226ce91ded8a16df828be5f566c013626c8ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 23:00:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"95c5c88b67e5f1184336339e28dbe6fb5cd23963","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 23:01:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbb61b0a464241a38338b0730ed95b7f\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/a5c0522b6e4c45c2b05dd8614d4874ec : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4ef299381baa78e300e754c13bb0be5022cfa8d9","revisions":{"69d320fd2e46ebab2cdd5ea5d9b2a09129e49508":{"kind":"REWORK","_number":1,"created":"2022-10-21 16:31:17.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/78/862378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/78/862378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/78/862378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/78/862378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/78/862378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/78/862378/1"}}},"commit":{"parents":[{"commit":"8ae13def4d01524487a2b05d5b07cea3cc676f9d","subject":"Antelope Election Results (TC/PTL)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8ae13def4d01524487a2b05d5b07cea3cc676f9d"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-18 16:03:42.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:24:42.000000000","tz":-300},"subject":"Switch election timing to Wednesday","message":"Switch election timing to Wednesday\n\nTuesday was selected because it was early in the week and was \"safe\"\nfor APAC and US timezones.   Releases actually happen on Wednesdays\nso for something more like consistency lets do make elections also\nhappen on Wednesdays\n\nChange-Id: If66602b12eb77b17904427149ab81ce7a21d9492\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/69d320fd2e46ebab2cdd5ea5d9b2a09129e49508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/69d320fd2e46ebab2cdd5ea5d9b2a09129e49508"}]},"branch":"refs/heads/master"},"6dbea55db68c70339d21a5020e0f4e0a01a7d9b4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-01 11:01:31.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/78/862378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/78/862378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/78/862378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/78/862378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/78/862378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/78/862378/2"}}},"commit":{"parents":[{"commit":"3d240fe5c18a1fd147b6d1fa897fb075ca57c11a","subject":"tox4 updates for editable package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3d240fe5c18a1fd147b6d1fa897fb075ca57c11a"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-18 16:03:42.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-01 11:01:15.000000000","tz":660},"subject":"Switch election timing to Wednesday","message":"Switch election timing to Wednesday\n\nTuesday was selected because it was early in the week and was \"safe\"\nfor APAC and US timezones.   Releases actually happen on Wednesdays\nso for something more like consistency lets do make elections also\nhappen on Wednesdays\n\nChange-Id: If66602b12eb77b17904427149ab81ce7a21d9492\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6dbea55db68c70339d21a5020e0f4e0a01a7d9b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6dbea55db68c70339d21a5020e0f4e0a01a7d9b4"}]},"branch":"refs/heads/master"},"4ef299381baa78e300e754c13bb0be5022cfa8d9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-01 22:24:25.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/78/862378/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/78/862378/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/78/862378/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/78/862378/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/78/862378/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/78/862378/3"}}},"commit":{"parents":[{"commit":"e9aba065ff0868da21558cbeddcfea29fd644ea8","subject":"Merge \"Fix release name and timeline names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e9aba065ff0868da21558cbeddcfea29fd644ea8"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-18 16:03:42.000000000","tz":-300},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-02-01 22:23:45.000000000","tz":540},"subject":"Switch election timing to Wednesday","message":"Switch election timing to Wednesday\n\nTuesday was selected because it was early in the week and was \"safe\"\nfor APAC and US timezones.   Releases actually happen on Wednesdays\nso for something more like consistency lets do make elections also\nhappen on Wednesdays\n\nChange-Id: If66602b12eb77b17904427149ab81ce7a21d9492\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4ef299381baa78e300e754c13bb0be5022cfa8d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4ef299381baa78e300e754c13bb0be5022cfa8d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[]}
