)]}'
{"id":"openstack%2Felection~862379","triplet_id":"openstack%2Felection~master~I9f9851c76455e5d559ba68921dfd29557a64b574","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-02-02 00:50:11.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-02-02 01:03:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9f9851c76455e5d559ba68921dfd29557a64b574","subject":"Remove stupid think-o","status":"MERGED","created":"2022-10-21 16:31:17.000000000","updated":"2023-02-02 01:03:54.000000000","submitted":"2023-02-02 01:03:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862379","meta_rev_id":"550b30ddb38c03125931efb204f2ca2bec5e26ae","_number":862379,"virtual_id_number":862379,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-02 01:03:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-02-02 00:39:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-02-02 00:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-02 00:39:02.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2023-02-02 00:39:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2023-02-02 00:50:11.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 16:32:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-10-21 16:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 00:55:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-11-23 16:45:27.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-12-06 20:33:25.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"a83bed40453c990c17f83a6831ac1c31d0c19f5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-10-21 16:31:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76b175b0345221689427880ea1f9ee7789663412","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-10-21 16:32:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"521ebd586de3f96dcc72dbd70ab76169798beac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:40:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9353bddc0b1440ccbbcd6a0b51c687d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8dfe37252140e5b9cc6b5e0ef2867c : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90e107172659435b869cc3bbf68ce1e8 : SUCCESS in 3m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0cce28d0ac884381b0822f74b9eb83a9 : SUCCESS in 2m 40s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2168cea91cdf47e99659a2675607a8b5 : SUCCESS in 2m 48s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a101712b3a25417caedf0edc89849bf0 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1632a3a03046d0763e3c2e7ea476292a753ce4d6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-23 00:55:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"75a0cd2d4a721f0c9f9e3418db787eced543ab20","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-23 16:45:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8a1c8e53eb7135ce329622734a4b48232ef458e3","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-12-06 20:33:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"514857662840deb6bff45d8392782cf53b5c702c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-31 22:12:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f95563e38c424890817327130237148d1a3af7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:12:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c90d65860146d0d28ef1adfbe5eaa10ba87da35d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:25:27.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/862378 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"68b8a4e686470514db9e93850714b2ea5f26b292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-01 11:01:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"10d42016502677363eab5822352e7183007c04a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:12:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/422e4602bcdc4284bd2c5df9f3cc05f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e762e102c9402eb21b815529ff12ca : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87625330fcac4ac7ab63873f16bf1324 : SUCCESS in 3m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/80d682f629d54101ace1d1a7e83570e7 : SUCCESS in 2m 59s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7a526ba77dec4d4ca710654ac267e959 : SUCCESS in 3m 09s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a802d2d06b7141999a92ef260a44f23c : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"b72609e8d3bd3be9d6fb84029b8bbf3eb03b428b","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 14:24:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"187adc4958f4ce4208b7cd32af8e160c67e04f34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 23:12:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"fbdc00c70791a64e9eddedce224033f59e24caed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 23:30:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e05076b86c4799bd5fac6c7725469e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb3d7644dbc645a986eec11ddc0c6840 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98117c7084154cb6a3b032d34f9b2707 : SUCCESS in 3m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5d2a7c1f43d94493b28bba5ca1aa0c63 : SUCCESS in 3m 27s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/75df2925ecbd4d65b701067767f331db : SUCCESS in 2m 53s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e8f2be6603fe431fbcad4099e0fe9597 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"021ac4a25e826afe6d61598fa47bd90db1b15d4b","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 23:47:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3d2e58c8181542aaae1955e4b4ab113bd05829a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 23:47:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9256c8c699d6bc7bece8c4c2cc3b15d348a2852a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-01 23:50:42.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3130906054115aa15de96fc6ff318dbee190fb57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:03:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d731132944194fd29a9ed539a7d21f51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a6e166dd0540178baf84b31a51e3d6 : FAILURE in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b1ca008ed414485ad1c54c4a03beb5b : SUCCESS in 2m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6b8aade998646709fc0ded6004243b9 : SUCCESS in 2m 58s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/3cd488ac52b34728b42996bbd9e89cc3 : FAILURE in 2m 51s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/914d964d5e724911992cfeba7c9fa452 : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":4},{"id":"1d199d820110fc5883fa7cd85c6e117afcf1410d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:12:51.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"78eda24ff23d8aab0f21ed05e81f53df3de79a06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:21:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c827a9fd355475b94b8c07485505833\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3477c26089140318be0487bd61a6666 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e12e71e8803491897a2b6a7ad2e79d6 : SUCCESS in 3m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/946c99cb697546eea48b05add1622de7 : SUCCESS in 3m 05s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/730d57f15dd547cb86712c225eb5794e : FAILURE in 2m 40s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8a694e71f4944b5aa16e2e18d25695b6 : FAILURE in 2m 45s","accounts_in_message":[],"_revision_number":5},{"id":"a36e3a02bce3769a3cd5c5fbd14dc8e8aae4fd6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:39:02.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"553196391feba12b461504ec0468c2d41cc3a2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:47:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f94678362001457a8e314af8fd91f439\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080b36eeb6b54552becd12f619fe8929 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b17289e97f124c2196f1ceb48240192b : SUCCESS in 2m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/38d0a460e81f47aebf7eb51229af23ad : SUCCESS in 2m 45s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/5159cd3e715d47cdb2abb353832784a3 : SUCCESS in 3m 04s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8d91d41f8dc645a3af31318e8aecf173 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":6},{"id":"cdc6c605f3024e05b3e014a754401cffa9722ae7","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:50:11.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f96cc42a4080c8582e26d3792dc5d50161e2f39d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:50:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a343e0a2526c2ce25f10ac0f034445797705f311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 01:03:54.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a174720cf5b4aabbd0bfe318d274358\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d6381b23774a6e8230d2b0ea8d1939 : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eafaed2333c4829b9013cc85e8aadc7 : SUCCESS in 3m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1b794e59318d4dd9aee84d6fa86fad1e : SUCCESS in 3m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/211dafff4e51416589ded3e0dc84d55d : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":6},{"id":"550b30ddb38c03125931efb204f2ca2bec5e26ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 01:03:54.000000000","message":"Change has been successfully merged\n\n3 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bb587d0ff476ba69e3c7fbac73fb72b139fd988f","revisions":{"953f8a2a4d8ee4813a785232f4617dc1427477e7":{"kind":"REWORK","_number":1,"created":"2022-10-21 16:31:17.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/79/862379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/79/862379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/79/862379/1"}}},"commit":{"parents":[{"commit":"69d320fd2e46ebab2cdd5ea5d9b2a09129e49508","subject":"Switch election timing to Wednesday","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/69d320fd2e46ebab2cdd5ea5d9b2a09129e49508"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-18 16:05:48.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:27:25.000000000","tz":-300},"subject":"Remove stupid think-o","message":"Remove stupid think-o\n\nSetting a variable to itself should be a no-op right?  Frankly I have\nno idea why I did this so remove it.\n\nChange-Id: I9f9851c76455e5d559ba68921dfd29557a64b574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/953f8a2a4d8ee4813a785232f4617dc1427477e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/953f8a2a4d8ee4813a785232f4617dc1427477e7"}]},"branch":"refs/heads/master"},"41412974cb2731b31aafd2aa4e890918b8f80b7a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-01 11:01:31.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/79/862379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/79/862379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/79/862379/2"}}},"commit":{"parents":[{"commit":"6dbea55db68c70339d21a5020e0f4e0a01a7d9b4","subject":"Switch election timing to Wednesday","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6dbea55db68c70339d21a5020e0f4e0a01a7d9b4"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-18 16:05:48.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-01 11:01:15.000000000","tz":660},"subject":"Remove stupid think-o","message":"Remove stupid think-o\n\nSetting a variable to itself should be a no-op right?  Frankly I have\nno idea why I did this so remove it.\n\nChange-Id: I9f9851c76455e5d559ba68921dfd29557a64b574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/41412974cb2731b31aafd2aa4e890918b8f80b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/41412974cb2731b31aafd2aa4e890918b8f80b7a"}]},"branch":"refs/heads/master"},"e09438aa75ee16cfabbddf2c98e31ccea3addd33":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-01 23:12:19.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/79/862379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/79/862379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/79/862379/3"}}},"commit":{"parents":[{"commit":"4ef299381baa78e300e754c13bb0be5022cfa8d9","subject":"Switch election timing to Wednesday","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4ef299381baa78e300e754c13bb0be5022cfa8d9"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-18 16:05:48.000000000","tz":-300},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-02-01 23:12:00.000000000","tz":540},"subject":"Remove stupid think-o","message":"Remove stupid think-o\n\nSetting a variable to itself should be a no-op right?  Frankly I have\nno idea why I did this so remove it.\n\nChange-Id: I9f9851c76455e5d559ba68921dfd29557a64b574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e09438aa75ee16cfabbddf2c98e31ccea3addd33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e09438aa75ee16cfabbddf2c98e31ccea3addd33"}]},"branch":"refs/heads/master"},"f2b17c420d92315d59c472784e51a26c39b49583":{"kind":"NO_CHANGE","_number":4,"created":"2023-02-01 23:50:42.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/79/862379/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/79/862379/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/79/862379/4"}}},"commit":{"parents":[{"commit":"4ef299381baa78e300e754c13bb0be5022cfa8d9","subject":"Switch election timing to Wednesday","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4ef299381baa78e300e754c13bb0be5022cfa8d9"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-18 16:05:48.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-01 23:50:05.000000000","tz":660},"subject":"Remove stupid think-o","message":"Remove stupid think-o\n\nSetting a variable to itself should be a no-op right?  Frankly I have\nno idea why I did this so remove it.\n\nChange-Id: I9f9851c76455e5d559ba68921dfd29557a64b574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f2b17c420d92315d59c472784e51a26c39b49583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f2b17c420d92315d59c472784e51a26c39b49583"}]},"branch":"refs/heads/master"},"75ba36c5921b67b21b07b8713689efe63e254e9e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-02-02 00:12:51.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/79/862379/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/79/862379/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/79/862379/5"}}},"commit":{"parents":[{"commit":"86fc45587c4cd1c41f817d79d68442a40cbc958f","subject":"fix release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/86fc45587c4cd1c41f817d79d68442a40cbc958f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-18 16:05:48.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-02 00:06:29.000000000","tz":660},"subject":"Remove stupid think-o","message":"Remove stupid think-o\n\nSetting a variable to itself should be a no-op right?  Frankly I have\nno idea why I did this so remove it.\n\nChange-Id: I9f9851c76455e5d559ba68921dfd29557a64b574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/75ba36c5921b67b21b07b8713689efe63e254e9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/75ba36c5921b67b21b07b8713689efe63e254e9e"}]},"branch":"refs/heads/master"},"bb587d0ff476ba69e3c7fbac73fb72b139fd988f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-02-02 00:39:02.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/79/862379/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/79/862379/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/79/862379/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/79/862379/6"}}},"commit":{"parents":[{"commit":"780c4140cb58f9cee947b5d79461c9ac52e9abea","subject":"fix release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/780c4140cb58f9cee947b5d79461c9ac52e9abea"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-18 16:05:48.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-02 00:30:00.000000000","tz":660},"subject":"Remove stupid think-o","message":"Remove stupid think-o\n\nSetting a variable to itself should be a no-op right?  Frankly I have\nno idea why I did this so remove it.\n\nChange-Id: I9f9851c76455e5d559ba68921dfd29557a64b574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/bb587d0ff476ba69e3c7fbac73fb72b139fd988f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/bb587d0ff476ba69e3c7fbac73fb72b139fd988f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
