)]}'
{"id":"openstack%2Felection~862381","triplet_id":"openstack%2Felection~master~I14a2d71a0bc26c837ebe4dfebbcfc9b05f02b069","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-02-02 00:50:32.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"I14a2d71a0bc26c837ebe4dfebbcfc9b05f02b069","subject":"Update select_release_end_date() to work cross release","status":"MERGED","created":"2022-10-21 16:31:17.000000000","updated":"2023-02-02 01:05:38.000000000","submitted":"2023-02-02 01:03:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862381","meta_rev_id":"95554a7189ded88e4cb125d79463e8dc10ae38c3","_number":862381,"virtual_id_number":862381,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-02 01:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2023-02-02 00:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-02 00:39:02.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2023-02-02 00:39:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2023-02-02 00:52:18.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 16:40:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 00:57:46.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-11-23 16:46:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-12-06 20:36:29.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"305292226dd2672df7b245f39c0f4f554d7a47b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-10-21 16:31:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"754dca2fad97218535cab10d5282a13c784896b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:40:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdb03d2b608d43cd90a85a9841c444b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13380119833e4c2b88d311cd89780936 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/883195819f714c7b8d883907b7b5d0fe : SUCCESS in 3m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/49177545495f4f3a80fabe3b88c86484 : SUCCESS in 2m 37s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/61282d5f9df440f985010dfb6f02ac82 : SUCCESS in 2m 37s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/99dd6cd138f244eb9adb618a371c7088 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8f484d6a5715ee7c9ed6e5c181cfd0f26b95037c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-23 00:57:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"41c9c2fdb20ed3c150550586583c9e69969a8369","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-23 16:46:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d3fc3b2b294ee9e800d32e4f72b2f75205fc3044","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-12-06 20:36:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acec3b8dd7dca694d5948c9a796ae1874719481d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-31 22:12:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6b92a728cae559323066c4fa4ea62953ac4db7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:13:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e492365411cff4f43bb5f2ec40271dcfd0511bdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:25:28.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/862380, https://review.opendev.org/862379, https://review.opendev.org/862378 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"e25566381eb4c72c203f3cd69b33f90ba2963004","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-01 11:01:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e902c244e37e1e8cd1105118844535b8e82a33a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:21:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/844b4d8fc0c94a36a7d18561f5c1b9ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d31d32f7c6447789f8cffa6dbac82cd : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/124c60db6d71482eba07d8cb50b75dd6 : SUCCESS in 3m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d49029957680425ca54e50d1c6cd7754 : SUCCESS in 2m 57s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d5c6a8d4c00e445d9b803d0e83e10f13 : SUCCESS in 3m 05s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/91d5ef3a809c45729f3f6d6f84814231 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"b6276569f20f39a48680c86f9ecf7c14145a5a94","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 14:23:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5dac1eeb8793dfa57cbd41fe2db522a1f0f11a6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-01 23:50:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"40202064f2cc75528713a47b6adc4c10e10e9fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:01:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00701de87c354df6b307d7d6ef5d41e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02211120664c45c6b7438e398d33fea5 : FAILURE in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46f1dce1ef10454f8460464f4a8f5d3d : SUCCESS in 3m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5488c97233848af9c9d7a50502055a1 : SUCCESS in 2m 39s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/65b063c703094df09a924ce3d26cc459 : FAILURE in 3m 14s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/53866ce60ae7461d81e7f5c73109ec28 : FAILURE in 2m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f8361240138078c0ccee9c64b6af26a27c5b5e88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:12:51.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"0e9a0ea460c005618e144c6924012551beb2d56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:24:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34067a7478224c508e3447eb23973022\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee59416f38de46aeb00548b71bc00cca : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aca750fe78645ee8ba650dcd303b05c : SUCCESS in 2m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/201793373fd449f491e00911e8d64151 : SUCCESS in 2m 57s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/fb56d662148b4c2e917ba055f170f838 : FAILURE in 2m 58s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/dc4b97912f3e49d8b75102848fc16d83 : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":4},{"id":"706fcde2a3d842860908c78dc0ad32480acea760","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:39:02.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"537c90f7965cfa36ad52b4cf4b5db92ca9805f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:47:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04409f31ac654ae1a1675f331ac1a0cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5496c2ebb746480b816deffeff56a114 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb717c642d344d9c92f21f751bd921f3 : SUCCESS in 2m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/66c305832bfa41cb965aacd624a23db3 : SUCCESS in 3m 00s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/1ff214b5e8504b19bdb56f421b047bdf : SUCCESS in 2m 46s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e444a90474084659a87f6f9d7b9a3ab3 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":5},{"id":"e795c6f5d5f1ef80288a9190d9ffe9f929fa1ca3","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:50:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"423677c8c6e9905a30ec35e555e5f1e0821ae78a","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:52:18.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f80736e916fe0ce65fc360318858ca8c646abd3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:52:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ef44abb664ab165e1392d7f6dea21712ce343702","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 01:03:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a2544dd0dbd49f1aa7e457a6718d659\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e60da9acd7d48ab8b9f28f562fc316d : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea8253f31d00458da3c9651465ca60c6 : SUCCESS in 3m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cc4d02ecc1174221a1c984baef344ac6 : SUCCESS in 3m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b1decf95f50f4f3ea089e8710c651273 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":5},{"id":"fb20eeefab2c6743262818d2552b291c7e65b4f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 01:03:58.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":5},{"id":"95554a7189ded88e4cb125d79463e8dc10ae38c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 01:05:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa84c7c6ae7d464da11f8b2a5334f21c\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/c7b57565cc3d4c51946129018b57927f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5701eea4412a9171c223b1b698913491eafc2115","revisions":{"3c9977bcc88e5299c91f4767d4eb03488937bfae":{"kind":"REWORK","_number":1,"created":"2022-10-21 16:31:17.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/81/862381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/862381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/862381/1"}}},"commit":{"parents":[{"commit":"7c4642a8eba2b90bebaedbf682c8fe43efd1726f","subject":"Gracefully exit if we can\u0027t find a date","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7c4642a8eba2b90bebaedbf682c8fe43efd1726f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:20:00.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:27:25.000000000","tz":-300},"subject":"Update select_release_end_date() to work cross release","message":"Update select_release_end_date() to work cross release\n\nCurrently the event name is limited to \u0027x-final\u0027, this doesn\u0027t\nwork for current releases.  Update that to determine the correct\ntag.\n\nChange-Id: I14a2d71a0bc26c837ebe4dfebbcfc9b05f02b069\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3c9977bcc88e5299c91f4767d4eb03488937bfae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3c9977bcc88e5299c91f4767d4eb03488937bfae"}]},"branch":"refs/heads/master"},"e7b564c8aa1016a327fa77003ee32cfcb11a6f6e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-01 11:01:31.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/81/862381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/862381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/862381/2"}}},"commit":{"parents":[{"commit":"59af06c3a9c893d892e95899d99c815f1508ce14","subject":"Gracefully exit if we can\u0027t find a date","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/59af06c3a9c893d892e95899d99c815f1508ce14"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:20:00.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-01 11:01:15.000000000","tz":660},"subject":"Update select_release_end_date() to work cross release","message":"Update select_release_end_date() to work cross release\n\nCurrently the event name is limited to \u0027x-final\u0027, this doesn\u0027t\nwork for current releases.  Update that to determine the correct\ntag.\n\nChange-Id: I14a2d71a0bc26c837ebe4dfebbcfc9b05f02b069\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e7b564c8aa1016a327fa77003ee32cfcb11a6f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e7b564c8aa1016a327fa77003ee32cfcb11a6f6e"}]},"branch":"refs/heads/master"},"0a6a9ffd5f0ea3244669e66deab062fd1ed4e5b9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-01 23:50:42.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/81/862381/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/862381/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/862381/3"}}},"commit":{"parents":[{"commit":"8e9af1ba16e5089bee3c327f07db0bc8e62d3ffe","subject":"Gracefully exit if we can\u0027t find a date","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8e9af1ba16e5089bee3c327f07db0bc8e62d3ffe"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:20:00.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-01 23:50:05.000000000","tz":660},"subject":"Update select_release_end_date() to work cross release","message":"Update select_release_end_date() to work cross release\n\nCurrently the event name is limited to \u0027x-final\u0027, this doesn\u0027t\nwork for current releases.  Update that to determine the correct\ntag.\n\nChange-Id: I14a2d71a0bc26c837ebe4dfebbcfc9b05f02b069\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0a6a9ffd5f0ea3244669e66deab062fd1ed4e5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0a6a9ffd5f0ea3244669e66deab062fd1ed4e5b9"}]},"branch":"refs/heads/master"},"acc6540b88c9c1f8753c2f26761ae763a38159f1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-02 00:12:51.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/81/862381/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/862381/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/862381/4"}}},"commit":{"parents":[{"commit":"7a4d91d827504ca3580d761c11fb56f59f0b3e5f","subject":"Gracefully exit if we can\u0027t find a date","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7a4d91d827504ca3580d761c11fb56f59f0b3e5f"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:20:00.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-02 00:07:55.000000000","tz":660},"subject":"Update select_release_end_date() to work cross release","message":"Update select_release_end_date() to work cross release\n\nCurrently the event name is limited to \u0027x-final\u0027, this doesn\u0027t\nwork for current releases.  Update that to determine the correct\ntag.\n\nChange-Id: I14a2d71a0bc26c837ebe4dfebbcfc9b05f02b069\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/acc6540b88c9c1f8753c2f26761ae763a38159f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/acc6540b88c9c1f8753c2f26761ae763a38159f1"}]},"branch":"refs/heads/master"},"5701eea4412a9171c223b1b698913491eafc2115":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-02-02 00:39:02.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/81/862381/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/862381/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/862381/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/862381/5"}}},"commit":{"parents":[{"commit":"d084748d4456a05abf73a0c380a0a5494ad5715a","subject":"Gracefully exit if we can\u0027t find a date","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d084748d4456a05abf73a0c380a0a5494ad5715a"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:20:00.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-02 00:32:18.000000000","tz":660},"subject":"Update select_release_end_date() to work cross release","message":"Update select_release_end_date() to work cross release\n\nCurrently the event name is limited to \u0027x-final\u0027, this doesn\u0027t\nwork for current releases.  Update that to determine the correct\ntag.\n\nChange-Id: I14a2d71a0bc26c837ebe4dfebbcfc9b05f02b069\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5701eea4412a9171c223b1b698913491eafc2115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5701eea4412a9171c223b1b698913491eafc2115"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
