)]}'
{"id":"openstack%2Felection~862382","triplet_id":"openstack%2Felection~master~I71b65b82e8609fc4b3b48ef58acdf434ce69fe5c","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-02-02 00:50:46.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"I71b65b82e8609fc4b3b48ef58acdf434ce69fe5c","subject":"Simplify date conversions","status":"MERGED","created":"2022-10-21 16:31:17.000000000","updated":"2023-02-02 01:07:12.000000000","submitted":"2023-02-02 01:06:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862382","meta_rev_id":"f4318e0d2e461fc11c75b1e8deb65d59321da96d","_number":862382,"virtual_id_number":862382,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-02 01:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-02-02 00:39:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-02 00:39:02.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2023-02-02 00:39:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2023-02-02 00:50:46.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 16:49:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 00:58:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-11-23 16:47:09.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-12-06 20:36:11.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"0de2fb91f5a30bea788217bb1c218da0a9fadfdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-10-21 16:31:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c74c9d58d94ece42bda5f818f2a90a7e7facfd1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:49:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdfbbdd53c134bb0b54e7b8b19606012\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc816030dd04481e94fd885bdaa1d253 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/110e9ffb67d94d1995d02fae756825fc : SUCCESS in 3m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/772d9219ae4a4c96b9e700908b26b549 : SUCCESS in 3m 05s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8ffe51d0ab29437f988a5b03a27c0f0a : SUCCESS in 2m 49s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/bcda837ae6c64e81bbd53342e80bb497 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"634c03ff221d40966505306d84a215de57a630b5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-23 00:58:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"31cecc5d3bd93f35a94aae73482fae30cbe4caf4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-23 16:47:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2873e1aa083789e2e0d9c2783c690f3a76f595e","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2022-12-06 20:36:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a7d8c62ecfabf0ec12ee9b1a6b3bed98a6aaafe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-31 22:13:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b9afeacbf1b608abab37ff1e4053c10e29d0098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:13:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8c2d7b4e327a9d6344e0afef0d18b61e80169cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:25:29.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/862381, https://review.opendev.org/862380, https://review.opendev.org/862379, https://review.opendev.org/862378 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"50ef715e3d899311662b6eb6b6f251c723cad358","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-01 11:01:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d994b6bec562c26ab883dd0a008e1780cce15f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:20:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73b5f87ca6e14f7b95675b3407f24412\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816478f816484b5fb7293a0eafd7e618 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d37a259c7c374e7ab90c53db158e88c1 : SUCCESS in 4m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fed1a902f30945a29e24e19df04a26e2 : SUCCESS in 3m 10s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/5247179c45984960adac1ba24f659d96 : SUCCESS in 3m 32s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f0807e2a146749299625b316117cca8d : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"c90b0d34b3dd9eda1107653b4a4c3dfab7495b22","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 14:24:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3b9556906547fe65bf695fbfdfa8dda67cda6753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-01 23:50:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"66c13ded7fb43058f134d0dac329205f963c8f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 23:59:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/493cc7c3b4a243799fd5a6ee73b6eec9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b6de709b7a2458bad1654c0519a3ba1 : FAILURE in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28870be438d44c3bb75c53cc0aa2c462 : SUCCESS in 2m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d3edbb6c9c9644d38158549f1cf2fa0b : SUCCESS in 2m 41s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/88ce9d0c45d2455eaae8739797c91b12 : FAILURE in 2m 44s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/49bfc6d40c99419e8cc1c20ceb54888b : FAILURE in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"0758ffe0e52678b7e81a3502e225019a51421517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:12:51.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ceb003b4a097c593c505c72252712dec2856b339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:24:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8720e49202ac49a3bfcf04ed5f56946c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be06f2a32f3f4b38bd877c3ceace4a40 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b4b257ab5394780b8d044e6f2795869 : SUCCESS in 2m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3f3b197452674aa0a3a3c6597bc8ef8e : SUCCESS in 2m 53s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/92364f9198934001b2b40a787457d142 : FAILURE in 2m 51s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/198987fbb4334d1c8739ab7435b667ad : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":4},{"id":"cf12198e27a080bbd23b16794baeea0a65d0c654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:39:02.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"360f36c7be87ef14b577e92fc5980177465395e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:48:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c027b43a2cb34ffd97b24f6a667e6cc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ae5aac017744bb185a7269109e8c3b8 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/317b2bd2c6c94a7ca77cc3cb9e9bc16f : SUCCESS in 2m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/98fd3c5ee0ec49fca9657ba1488a65b7 : SUCCESS in 2m 45s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/be1696fd65874e8b998dbd2bb3b4c2c5 : SUCCESS in 4m 13s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7e21d1727f274c64862e58899b9a785e : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":5},{"id":"6c4e4bb9b606b7af91f51712444690867cac5501","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 00:50:46.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cb73e112cb03db0a7487b83c5e94865cee8b21ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 00:52:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ac5badf6eb774a4a73655f301afadcc9b599c6b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 01:06:15.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e748c115fef427990a4fc1084de45ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093eec5ac3df4e0fac49f1a73b7c9ddf : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d03793fbea4c43cbbcd5be22d471d491 : SUCCESS in 3m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6b676b1c8a04362b6da93ec5f137a7e : SUCCESS in 3m 17s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9dbbcac5481e49659d4be1e3832c266a : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":5},{"id":"a0091e604829dc8b523665eebb32dd4c767e5bb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 01:06:16.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":5},{"id":"f4318e0d2e461fc11c75b1e8deb65d59321da96d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 01:07:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15fbad11c21d49d2b5f92b9d4e418c74\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/9b0f75c6ee354b4cb986e6a9eae3242d : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"84c50c9878ab051ffd4a5fce20e8f544bbd032aa","revisions":{"d485e6027cbc21187af45bdc8537ee12e0b62163":{"kind":"REWORK","_number":1,"created":"2022-10-21 16:31:17.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/82/862382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/82/862382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/82/862382/1"}}},"commit":{"parents":[{"commit":"3c9977bcc88e5299c91f4767d4eb03488937bfae","subject":"Update select_release_end_date() to work cross release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3c9977bcc88e5299c91f4767d4eb03488937bfae"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:25:03.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:27:56.000000000","tz":-300},"subject":"Simplify date conversions","message":"Simplify date conversions\n\nAlso do a sneaky update to switch from dict() to list()\n\nChange-Id: I71b65b82e8609fc4b3b48ef58acdf434ce69fe5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d485e6027cbc21187af45bdc8537ee12e0b62163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d485e6027cbc21187af45bdc8537ee12e0b62163"}]},"branch":"refs/heads/master"},"0e258d876da2960fd094db3c6144e825874241a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-01 11:01:31.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/82/862382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/82/862382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/82/862382/2"}}},"commit":{"parents":[{"commit":"e7b564c8aa1016a327fa77003ee32cfcb11a6f6e","subject":"Update select_release_end_date() to work cross release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e7b564c8aa1016a327fa77003ee32cfcb11a6f6e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:25:03.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-01 11:01:15.000000000","tz":660},"subject":"Simplify date conversions","message":"Simplify date conversions\n\nAlso do a sneaky update to switch from dict() to list()\n\nChange-Id: I71b65b82e8609fc4b3b48ef58acdf434ce69fe5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0e258d876da2960fd094db3c6144e825874241a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0e258d876da2960fd094db3c6144e825874241a2"}]},"branch":"refs/heads/master"},"e7a1311ba00142fe0bcb9b934099b71f9f7944ad":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-01 23:50:42.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/82/862382/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/82/862382/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/82/862382/3"}}},"commit":{"parents":[{"commit":"0a6a9ffd5f0ea3244669e66deab062fd1ed4e5b9","subject":"Update select_release_end_date() to work cross release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0a6a9ffd5f0ea3244669e66deab062fd1ed4e5b9"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:25:03.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-01 23:50:05.000000000","tz":660},"subject":"Simplify date conversions","message":"Simplify date conversions\n\nAlso do a sneaky update to switch from dict() to list()\n\nChange-Id: I71b65b82e8609fc4b3b48ef58acdf434ce69fe5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e7a1311ba00142fe0bcb9b934099b71f9f7944ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e7a1311ba00142fe0bcb9b934099b71f9f7944ad"}]},"branch":"refs/heads/master"},"a9aaee6c26bd795c9a8403465e4a500a2efb9e80":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-02 00:12:51.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/82/862382/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/82/862382/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/82/862382/4"}}},"commit":{"parents":[{"commit":"acc6540b88c9c1f8753c2f26761ae763a38159f1","subject":"Update select_release_end_date() to work cross release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/acc6540b88c9c1f8753c2f26761ae763a38159f1"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:25:03.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-02 00:08:36.000000000","tz":660},"subject":"Simplify date conversions","message":"Simplify date conversions\n\nAlso do a sneaky update to switch from dict() to list()\n\nChange-Id: I71b65b82e8609fc4b3b48ef58acdf434ce69fe5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a9aaee6c26bd795c9a8403465e4a500a2efb9e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a9aaee6c26bd795c9a8403465e4a500a2efb9e80"}]},"branch":"refs/heads/master"},"84c50c9878ab051ffd4a5fce20e8f544bbd032aa":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-02-02 00:39:02.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/82/862382/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/82/862382/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/82/862382/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/82/862382/5"}}},"commit":{"parents":[{"commit":"5701eea4412a9171c223b1b698913491eafc2115","subject":"Update select_release_end_date() to work cross release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5701eea4412a9171c223b1b698913491eafc2115"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-21 16:25:03.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-02 00:33:23.000000000","tz":660},"subject":"Simplify date conversions","message":"Simplify date conversions\n\nAlso do a sneaky update to switch from dict() to list()\n\nChange-Id: I71b65b82e8609fc4b3b48ef58acdf434ce69fe5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/84c50c9878ab051ffd4a5fce20e8f544bbd032aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/84c50c9878ab051ffd4a5fce20e8f544bbd032aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
