)]}'
{"id":"openstack%2Felection~872231","triplet_id":"openstack%2Felection~master~If70b1a8b17bd925b04b35366ff80fd5dacf19c3c","project":"openstack/election","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-02-01 21:21:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If70b1a8b17bd925b04b35366ff80fd5dacf19c3c","subject":"Fix tox4 error","status":"MERGED","created":"2023-01-30 22:42:36.000000000","updated":"2023-02-01 21:22:51.000000000","submitted":"2023-02-01 21:21:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872231-tox4","meta_rev_id":"9f47c9a58cf4f93e9a1d92f16eb7346c4558c55e","_number":872231,"virtual_id_number":872231,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-01 21:21:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-01-30 23:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-30 22:53:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2023-02-01 21:17:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 22:51:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-30 22:53:23.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-01-30 23:01:44.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"ee6c12a1e668d29c8e9fb214a7c4199aca51657d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-30 22:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe53c25ffb37ef440de5047e5db2e709fdaea44f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-30 22:42:45.000000000","message":"Topic set to tox4","accounts_in_message":[],"_revision_number":1},{"id":"c0724794630e617082103cb57004f3170a61c66b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 22:51:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b8816feccfd4866b14ad70a01c9645b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b61b46183a44da2920c261c02245822 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7896245d7fa45adb19b3aad0d56ad1e : SUCCESS in 2m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f08c4ec8920b40a3b0dfe563e173ef30 : SUCCESS in 3m 21s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/018ddbd90bb7444fa9dbdf8604081cde : SUCCESS in 2m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2b16a4f342164e5e93a5b9a2bb48f249 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a0718a4be8d5207d65c3be29af5f10d691f360a8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-30 22:53:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"93c3df83fa43f0c8bbf5b25dd1ab3ba9b0e873d5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-01-30 23:01:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ae180b8b4c250eb2045476004da631689ac87a17","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-01 21:17:21.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82fc4c709a02839200625351c29b71617e14fb3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:17:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"765c907d2c02e92b602ebda16276fc47202d9c06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:21:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8ba7ab86ade4ffea437f3371f2b7bb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f6a4bf175f4343a8ffe7ecc463711c : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0791b083ecfd4f9e9da797df23d8dc6d : SUCCESS in 3m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3f8ac206575349fea7cebf28d8c4a108 : SUCCESS in 2m 34s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/84f6e17d16214de6a6675be7b83a36d7 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"35c7fcd9af742343411edefec00b8b38c57ab670","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:21:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9f47c9a58cf4f93e9a1d92f16eb7346c4558c55e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:22:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15b7611a3baf4b4c9a4b8376a6d1451a\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/47d58bf916f24479afb42ddda98c6a22 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fac3d836bc3a2660691ad909147d2b06220aa85e","revisions":{"fac3d836bc3a2660691ad909147d2b06220aa85e":{"kind":"REWORK","_number":1,"created":"2023-01-30 22:42:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/31/872231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/31/872231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/31/872231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/31/872231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/31/872231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/31/872231/1"}}},"commit":{"parents":[{"commit":"dac837c5966df1bfcf969a7c065c4754cc39c01e","subject":"Merge \"Add dates for the 2023.2/B-release[1]\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dac837c5966df1bfcf969a7c065c4754cc39c01e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-30 22:40:10.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-30 22:40:10.000000000","tz":-360},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One of the reason for failure\nis due to the skipsdist \u003d True with usedevelop and it\nseems it was never supported.\n\nChange-Id: If70b1a8b17bd925b04b35366ff80fd5dacf19c3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fac3d836bc3a2660691ad909147d2b06220aa85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fac3d836bc3a2660691ad909147d2b06220aa85e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[]}
