)]}'
{"id":"openstack%2Felection~872510","triplet_id":"openstack%2Felection~master~Icbdca2f72c0777c587d6f95b3fab1911be179f97","project":"openstack/election","branch":"master","topic":"setuptools-fix","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-03-24 22:00:38.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-03-24 21:56:23.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Icbdca2f72c0777c587d6f95b3fab1911be179f97","subject":"Update package metadata and tox4 setup","status":"MERGED","created":"2023-02-02 05:40:33.000000000","updated":"2023-03-24 22:01:27.000000000","submitted":"2023-03-24 22:00:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872510-setuptools-fix","meta_rev_id":"c6a8b3310d8bdc0c78abb0a791b849f1e81b8816","_number":872510,"virtual_id_number":872510,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-24 22:00:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-02-24 00:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2023-03-24 21:56:23.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2023-03-24 21:56:23.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-02 05:59:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 00:19:26.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2023-02-24 00:19:44.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"9b9e1ce9be78e1255395e6a559c732b0c1582fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 05:40:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a39c1ef6fa3646ebfb4328bd4073f501d3655c62","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-02 05:40:52.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"53d69afc988a0f38c10394d6a767d319f613ea8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 05:59:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/538a536d0ee84469915cd26f2d7ae866\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5dd17ec1abe4776b578fe2e727ad033 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cadf12b550f44353b7d171eab39b4cf5 : SUCCESS in 5m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54c9e6bedb5c47729637a30bb433b0f3 : SUCCESS in 3m 40s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/ebc91638c1f84e45a741294e2004c277 : SUCCESS in 2m 45s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/4d600790daf84462a0bbbd5c13f4cd27 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ddf04f3329b0950726640eccc6289ceed01e7097","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-22 03:37:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0c86738f4e283c0850cf74f0dda3d87d16be5461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 03:47:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b6ae4129c754038840d542876628f46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962c0d258b904f0e909036a1e5e5be48 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbd4a18b59b242eca51ffd85c2ea533c : SUCCESS in 3m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/49147e5e64b94c268864a830202c6a44 : SUCCESS in 3m 36s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/c78093522cc347ebbb7d5109c561f0bf : SUCCESS in 3m 02s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8ff8b431e0724722ab03fe2b2df1520d : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1188c2b4a59ef7b3ff7863e9996920c5e301c4f5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-24 00:19:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f3e5363b60853417e7deb4f695e486834c170a5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-24 21:56:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e8266d3dc16096776ee791225479d872452b42e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 21:56:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b29c3eb95122a9ee551993a31c933f02b7e7ea4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 22:00:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026211f1b75b44bd9104dcdc31812379\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf55a3ff7034cbcb67f4b56598ee89e : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16f3687b98744e6d92d781b23b1788cf : SUCCESS in 3m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1210df97c9984c49b0dafe452817c902 : SUCCESS in 2m 27s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7e8847e963c443f1b1e5775fa0a86bee : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":2},{"id":"bde4af9efe4142f8e7aa10760a8e184c5b2c55db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 22:00:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c6a8b3310d8bdc0c78abb0a791b849f1e81b8816","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 22:01:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4775d583b8394ee4a0894ddfbc3755fa\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/46dd07cea8b84b4ebd79c7ebff01fbc1 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"edfebf3dcc62711f1197ac9490bd1b9f6f38a6a2","revisions":{"949f606431a46f9f371aedec6949d2f9a279b7a9":{"kind":"REWORK","_number":1,"created":"2023-02-02 05:40:33.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/10/872510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/10/872510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/10/872510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/10/872510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/10/872510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/10/872510/1"}}},"commit":{"parents":[{"commit":"e8ea3abdd5f05eae5a5136dcbcbc7e8755f733ac","subject":"Merge \"Skip if e-mail in Gerrit is None\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e8ea3abdd5f05eae5a5136dcbcbc7e8755f733ac"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-02 05:36:10.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-02 05:36:10.000000000","tz":660},"subject":"Update package metadata and tox4 setup","message":"Update package metadata and tox4 setup\n\nThe current tox setup results in an \u0027editable-wheel\u0027 install which,\nwhile it works, generates unhelpful warnings.  Switch to the \u0027editable\u0027\ntype which is what you got under tox \u003c\u003d 4.\n\nWhile we are reducing noise in the logs, correctly setup\nopenstack_elections.templates as a module.  We need it as a module as\nwe rely on Jinja2 to be able to use the module loader to locate the\ntemplates.\n\nChange-Id: Icbdca2f72c0777c587d6f95b3fab1911be179f97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/949f606431a46f9f371aedec6949d2f9a279b7a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/949f606431a46f9f371aedec6949d2f9a279b7a9"}]},"branch":"refs/heads/master"},"edfebf3dcc62711f1197ac9490bd1b9f6f38a6a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-22 03:37:10.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/10/872510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/10/872510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/10/872510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/10/872510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/10/872510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/10/872510/2"}}},"commit":{"parents":[{"commit":"80f2ccfd426bf4bbd1a90c9e4758fbbb7dbeaef3","subject":"Adding Dave Wilde candidacy for the Keystone project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/80f2ccfd426bf4bbd1a90c9e4758fbbb7dbeaef3"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-02 05:36:10.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-02-22 03:37:10.000000000","tz":0},"subject":"Update package metadata and tox4 setup","message":"Update package metadata and tox4 setup\n\nThe current tox setup results in an \u0027editable-wheel\u0027 install which,\nwhile it works, generates unhelpful warnings.  Switch to the \u0027editable\u0027\ntype which is what you got under tox \u003c\u003d 4.\n\nWhile we are reducing noise in the logs, correctly setup\nopenstack_elections.templates as a module.  We need it as a module as\nwe rely on Jinja2 to be able to use the module loader to locate the\ntemplates.\n\nChange-Id: Icbdca2f72c0777c587d6f95b3fab1911be179f97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/edfebf3dcc62711f1197ac9490bd1b9f6f38a6a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/edfebf3dcc62711f1197ac9490bd1b9f6f38a6a2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
