)]}'
{"id":"openstack%2Felection~873495","triplet_id":"openstack%2Felection~master~Ia34a8680cf8f90703b7b3604e339e54df6804d4f","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-02-13 09:42:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia34a8680cf8f90703b7b3604e339e54df6804d4f","subject":"Add Vishal Manchanda candidacy for Horizon PTL(Bobcat)","status":"MERGED","created":"2023-02-13 08:04:12.000000000","updated":"2023-02-13 09:43:58.000000000","submitted":"2023-02-13 09:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873495","meta_rev_id":"3feaa9f09bdc61ead9889987d015401054ce606d","_number":873495,"virtual_id_number":873495,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-13 09:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2023-02-13 09:08:24.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2023-02-13 09:08:24.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 08:13:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 09:08:24.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"a8c40074b46895b339bd1633a1b09ad13d1b7400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-02-13 08:04:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"318d1f4d90c840b2ea73f53f1232ac8652105493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 08:13:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/242105c2e0d5468e867b7d8ccd933b37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b7177a2f3c4df9a2fa0450c05c865d : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc1801b860e54380ba595a98efc91588 : SUCCESS in 2m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/62c891f5107d4e90a0876bf05af3514b : SUCCESS in 2m 59s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/88dfea190d0d479f9c349e4d009e9e4c : SUCCESS in 2m 55s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/57c481a8c23e46388739b0a29a057a7e : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"65622d477001a454f6eb477384b00b1566a51721","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-13 09:08:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4450c18df22ad6de341cba17838ff1e4b6ff125","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 09:08:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f797148f7b84e0e728234549358701d94be9584f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 09:42:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0d69e236ee24b218fca89a2682e67b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0bc2a3752f42b09d2939d02a1f2aa9 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41e34559d49847e18522d4c585c0dbf7 : SUCCESS in 5m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a2f4853bbd254511be9e4a65496cc453 : SUCCESS in 3m 29s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f249919edf1a4d829e6805897e9c4070 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"4d0ef97e6946d334bdc1702cddb841bd0b308471","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 09:42:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3feaa9f09bdc61ead9889987d015401054ce606d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 09:43:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/026e7771284a4ead8ce19fbb72ec7181\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/3ab7b0c51fcd4879a289a2320bd6b37c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"339fccb2ebaee63c720ff2398c44a447c1ce0e0f","revisions":{"339fccb2ebaee63c720ff2398c44a447c1ce0e0f":{"kind":"REWORK","_number":1,"created":"2023-02-13 08:04:12.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/95/873495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/95/873495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/95/873495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/95/873495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/95/873495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/95/873495/1"}}},"commit":{"parents":[{"commit":"1d9c0178f627c69d7342076d43afb78dc66de87e","subject":"Merge \"Yasufumi Ogawa\u0027s candidacy for Tacker PTL Bobcat/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1d9c0178f627c69d7342076d43afb78dc66de87e"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-13 08:03:18.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-02-13 08:03:18.000000000","tz":330},"subject":"Add Vishal Manchanda candidacy for Horizon PTL(Bobcat)","message":"Add Vishal Manchanda candidacy for Horizon PTL(Bobcat)\n\nChange-Id: Ia34a8680cf8f90703b7b3604e339e54df6804d4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/339fccb2ebaee63c720ff2398c44a447c1ce0e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/339fccb2ebaee63c720ff2398c44a447c1ce0e0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
