)]}'
{"id":"openstack%2Felection~873514","triplet_id":"openstack%2Felection~master~I44fdb168590f8e0f46535cf1298fd93cad3d7142","project":"openstack/election","branch":"master","topic":"bauzas_nova_bobcat","attention_set":{},"removed_from_attention_set":{"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-02-13 15:51:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I44fdb168590f8e0f46535cf1298fd93cad3d7142","subject":"Sylvain Bauza candidacy for Nova Bobcat PTL","status":"MERGED","created":"2023-02-13 09:59:45.000000000","updated":"2023-02-13 15:52:56.000000000","submitted":"2023-02-13 15:51:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873514-bauzas_nova_bobcat","meta_rev_id":"39a5c2c56551876ac9a00cd4a9bd0b2194efbfda","_number":873514,"virtual_id_number":873514,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-13 15:51:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-02-13 15:41:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2023-02-13 15:41:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 10:39:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 15:41:15.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"5cfa9f228dbefd8b3a40a7ceb414912d932aa525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-13 09:59:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"894c1fe8c6e612fba78a81d01fb4f73842865f1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 10:39:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f8d239c24314575b17a0adaa8385716\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd27a4b87bd4049971125303e9b7f06 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27e7257327ee44e1b7206a2e7a55d537 : SUCCESS in 3m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99bc5bde4eea4d42b67af7a815110673 : SUCCESS in 3m 22s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/44e399965b4e4f44b397dda5dd88b263 : SUCCESS in 4m 11s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5cfd4709e11040668fe62c564d48f509 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f732e9b4caef0c5206c03362ff9171550c370505","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-13 15:41:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a10a04a21caa211aeaadb04ad09004a907a4f7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:41:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"434b2f35d07acdee66c79024d3d62065118a76b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:51:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd4ce38488af42d5bb372b3b6f413e53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9518ab8bf8434c4da5e4a0fdb9614746 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41d92a7663894605959877ee8cc2d0a3 : SUCCESS in 3m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d1a46e6bc31c4969b1afc136f262d5e3 : SUCCESS in 3m 05s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ab44bc2f348b468e958a33c89b2bd541 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2653f3328b0e6ec28d1f8b874bfbc010b819182b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:51:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"39a5c2c56551876ac9a00cd4a9bd0b2194efbfda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:52:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70ffc2bef09548a69068ed1d73bd6199\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/4658108cd3464db7a4b6e9d878950298 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c19bd2f3a3161ecd36665d6d24d9b9efba54050c","revisions":{"c19bd2f3a3161ecd36665d6d24d9b9efba54050c":{"kind":"REWORK","_number":1,"created":"2023-02-13 09:59:45.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/14/873514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/14/873514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/14/873514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/14/873514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/14/873514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/14/873514/1"}}},"commit":{"parents":[{"commit":"1d9c0178f627c69d7342076d43afb78dc66de87e","subject":"Merge \"Yasufumi Ogawa\u0027s candidacy for Tacker PTL Bobcat/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1d9c0178f627c69d7342076d43afb78dc66de87e"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-13 09:59:07.000000000","tz":60},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2023-02-13 09:59:07.000000000","tz":60},"subject":"Sylvain Bauza candidacy for Nova Bobcat PTL","message":"Sylvain Bauza candidacy for Nova Bobcat PTL\n\nChange-Id: I44fdb168590f8e0f46535cf1298fd93cad3d7142\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c19bd2f3a3161ecd36665d6d24d9b9efba54050c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c19bd2f3a3161ecd36665d6d24d9b9efba54050c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[]}
