)]}'
{"id":"openstack%2Felection~873631","triplet_id":"openstack%2Felection~master~I3fa381ee9690f17eb199279659ea65f916610660","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-02-14 00:12:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fa381ee9690f17eb199279659ea65f916610660","subject":"Adding Michael Johnson candidacy for Designate","status":"MERGED","created":"2023-02-13 17:24:58.000000000","updated":"2023-02-14 00:14:34.000000000","submitted":"2023-02-14 00:12:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873631","meta_rev_id":"9262e7c7298467ad013ff1a1812b1930cc1e294d","_number":873631,"virtual_id_number":873631,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-14 00:12:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2023-02-13 23:57:18.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2023-02-13 23:57:18.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 17:40:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 23:57:18.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"160337e039912ae0de696978a8b15ddc7dcd4b98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-13 17:24:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e19d560f0ec3a38ec719886d0761a519e1fbb536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 17:40:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba394fc08e7e417c8dca97aad4658853\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4995907f444a5b91882ecde63f9bf8 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12e0e34275724cfd9611d655792b3e80 : SUCCESS in 2m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/46530b972c064517b178c71ddce493d1 : SUCCESS in 2m 56s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/3eec5b955bcf4ad89106290d5a72c258 : SUCCESS in 2m 37s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7b66e67d24594b1e86ddaf63d76457f6 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d2c3bbfe8975cf120a99ec6338d1a2f659881be8","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-13 23:57:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff55986050b3d837b6625039787ef6cf8f869a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 23:57:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9361f2099f22ec632c7750778ab751197d709b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 00:12:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b066de521f140549b1312b5056171f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8e87441ac246ac9e05c0cd540c4d1c : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d819ecd690154db19da620d760b3ac14 : SUCCESS in 3m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97c3f50bf0754c16978891885648d5d1 : SUCCESS in 2m 49s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/12d005f5bb474eb8aa146361dec0191b : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ec6d6916b975d3e2ecd59a2dcbb47bdb7d7b6d47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 00:12:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9262e7c7298467ad013ff1a1812b1930cc1e294d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 00:14:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ee3cbf8a3854470b608737ad3ee2d97\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/e039ba9e66e840c4acefe4337cd01c87 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed6bf52f1baa02253c4642cd6c127ab8aa1d792d","revisions":{"ed6bf52f1baa02253c4642cd6c127ab8aa1d792d":{"kind":"REWORK","_number":1,"created":"2023-02-13 17:24:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/31/873631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/31/873631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/31/873631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/31/873631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/31/873631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/31/873631/1"}}},"commit":{"parents":[{"commit":"e5d8a763a0700598109095e1e2830cfa26570811","subject":"Merge \"Sylvain Bauza candidacy for Nova Bobcat PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e5d8a763a0700598109095e1e2830cfa26570811"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-13 17:24:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-13 17:24:57.000000000","tz":0},"subject":"Adding Michael Johnson candidacy for Designate","message":"Adding Michael Johnson candidacy for Designate\n\nChange-Id: I3fa381ee9690f17eb199279659ea65f916610660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ed6bf52f1baa02253c4642cd6c127ab8aa1d792d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ed6bf52f1baa02253c4642cd6c127ab8aa1d792d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
