)]}'
{"id":"openstack%2Felection~873642","triplet_id":"openstack%2Felection~master~I3ba1e1424bd400404e3dbff857d0bed9b2ef510d","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-02-14 00:04:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ba1e1424bd400404e3dbff857d0bed9b2ef510d","subject":"Adding noonedeadpunk candidacy for OpenStack-Ansible PTL","status":"MERGED","created":"2023-02-13 19:55:44.000000000","updated":"2023-02-14 00:05:53.000000000","submitted":"2023-02-14 00:04:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873642","meta_rev_id":"db92855d0c53bade8c68dd2c778d359031fc0459","_number":873642,"virtual_id_number":873642,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-14 00:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2023-02-13 23:50:07.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2023-02-13 23:50:07.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 20:05:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 23:50:07.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"c79b1d10b3b619a83c06a3141173476b2026df98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-13 19:55:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14c70256ec7a645ed5184cc7fc2ca639e7d3ac2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 20:05:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41a9dd18915a4c68a9357e1c304f8599\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5957f4fbb242f79f96e2caa2fe6791 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96b7311c30584dc28f0452a101c97940 : SUCCESS in 3m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/06018fd23aa9497984c17d6fd32e3954 : SUCCESS in 2m 57s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6a54f046c596481baec2b3f6a8c88218 : SUCCESS in 3m 17s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2045eb57c82c4c928eb0d16269dfc469 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8a486dc724a74f9864bc73b2709745baa07832e0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-13 23:50:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"acf313ba0e7a91a3e679f146f440af0539e28601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 23:50:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab6d2b84245132a6ed5a91669e6fec947baf7161","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 00:04:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f028b4e360a34e839915c481b2e1c5e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ca7117713b435488bd523bec0f4162 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de6ad29725c147cfba1dd02dddc2904d : SUCCESS in 3m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/57f553b7c240480d88454e69107fac57 : SUCCESS in 2m 52s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/220b9849cde0426ab3d3b2646c936dbd : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"1b2e5f2c2481af245ea9868c3e2bca6366c85f85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 00:04:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db92855d0c53bade8c68dd2c778d359031fc0459","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 00:05:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/549731a65ea841f49702b591ff074591\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/602739b1bfe141ab837c00f4fdb507ff : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1b7b0642f086091fb8e58bfa8c942190dc6a6b8","revisions":{"d1b7b0642f086091fb8e58bfa8c942190dc6a6b8":{"kind":"REWORK","_number":1,"created":"2023-02-13 19:55:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/873642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/42/873642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/42/873642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/42/873642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/42/873642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/42/873642/1"}}},"commit":{"parents":[{"commit":"e5d8a763a0700598109095e1e2830cfa26570811","subject":"Merge \"Sylvain Bauza candidacy for Nova Bobcat PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e5d8a763a0700598109095e1e2830cfa26570811"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-02-13 19:55:35.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-02-13 19:55:35.000000000","tz":60},"subject":"Adding noonedeadpunk candidacy for OpenStack-Ansible PTL","message":"Adding noonedeadpunk candidacy for OpenStack-Ansible PTL\n\nChange-Id: I3ba1e1424bd400404e3dbff857d0bed9b2ef510d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d1b7b0642f086091fb8e58bfa8c942190dc6a6b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d1b7b0642f086091fb8e58bfa8c942190dc6a6b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
