)]}'
{"id":"openstack%2Felection~873756","triplet_id":"openstack%2Felection~master~I2d185dbc63312851cae2c4f4c4b4fd9821e07c1d","project":"openstack/election","branch":"master","topic":"tc-bobcat","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-02-15 02:26:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2d185dbc63312851cae2c4f4c4b4fd9821e07c1d","subject":"Add Slawek Kaplonski for TC in Bobcat","status":"MERGED","created":"2023-02-14 22:35:48.000000000","updated":"2023-02-15 02:27:41.000000000","submitted":"2023-02-15 02:26:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873756-tc-bobcat","meta_rev_id":"18da81b17e6cc140dc32c1152986e91e7e72b266","_number":873756,"virtual_id_number":873756,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-15 02:26:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2023-02-15 02:19:40.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2023-02-15 02:19:40.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-14 22:45:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-15 02:19:40.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"7731aef6e65feb069485bb713fe57908d1c5bd3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-02-14 22:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab9d8dfb0109a28fa9d2fa1e89b533a68fe24871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 22:45:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55500f249c894a0ea166d1d8e9a1f303\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3778ad843d39486c9d942abf392c02bb : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/821a3b7a1ebe42aab523b13d5365c1d9 : SUCCESS in 2m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a46d1798d442444594fa573a1b9a82cd : SUCCESS in 2m 52s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/74180a0ef33341f1b6c939549ab5e1e9 : SUCCESS in 2m 35s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5bdda9c75a6f4efb9e9dfba12d27603b : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"8a12200780cb1431907c0f56339002cd6c9bddbf","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-15 02:19:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d025d38aa01397f6ddee79367004fb962aa57941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 02:20:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0a2ea300090445c809cfa24091c69e19272859b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 02:26:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88c6af3ff9c24fb18327191819c9f67d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49703ff58cdc49be88b581b3fad6e1f2 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2ed09080cb74362b453a8f9857ee271 : SUCCESS in 3m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ee2cef7721674f79ba8781990e384097 : SUCCESS in 2m 53s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/39c50c97ebb545918a13df5b692986f8 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"56d4380d105bb43bf56d10baddcd3bb7354650b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 02:26:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"18da81b17e6cc140dc32c1152986e91e7e72b266","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 02:27:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/130044db0ae64477a52282fa7aabd2d1\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/0c3a1e390f1c4704ae4dd473a1006869 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"144cf78f5379e6931caa24d620edd1525023a311","revisions":{"144cf78f5379e6931caa24d620edd1525023a311":{"kind":"REWORK","_number":1,"created":"2023-02-14 22:35:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/56/873756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/56/873756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/56/873756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/56/873756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/56/873756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/56/873756/1"}}},"commit":{"parents":[{"commit":"0089cbeaeb4fde75c1fc7988cf1dbf62eaa8cb76","subject":"Merge \"Adding Roman Dobosz candidacy for Kuryr PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0089cbeaeb4fde75c1fc7988cf1dbf62eaa8cb76"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-02-14 22:32:09.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-02-14 22:35:46.000000000","tz":60},"subject":"Add Slawek Kaplonski for TC in Bobcat","message":"Add Slawek Kaplonski for TC in Bobcat\n\nChange-Id: I2d185dbc63312851cae2c4f4c4b4fd9821e07c1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/144cf78f5379e6931caa24d620edd1525023a311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/144cf78f5379e6931caa24d620edd1525023a311"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
