)]}'
{"id":"openstack%2Felection~873851","triplet_id":"openstack%2Felection~master~Ifbd5e3ac0a18e0492ea2b700d15462eaab676b02","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"35271":{"account":{"_account_id":35271,"name":"Shuai Qian","display_name":"qian.shuai","email":"qs_cloudnative@163.com","username":"iauhsnaiq"},"last_update":"2023-02-15 06:25:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifbd5e3ac0a18e0492ea2b700d15462eaab676b02","subject":"Add qian.shuai candidacy for Magnum PTL","status":"ABANDONED","created":"2023-02-15 03:45:20.000000000","updated":"2023-02-15 06:25:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ef1675b3ffdd5f631a7b0096a78f542d7d1ff2b8","_number":873851,"virtual_id_number":873851,"owner":{"_account_id":35271,"name":"Shuai Qian","display_name":"qian.shuai","email":"qs_cloudnative@163.com","username":"iauhsnaiq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-02-15 05:51:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 03:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25f322c06fcbb806ee5523f0f37de39c94bb86df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35271,"name":"Shuai Qian","display_name":"qian.shuai","email":"qs_cloudnative@163.com","username":"iauhsnaiq"},"date":"2023-02-15 03:45:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47a93ec3f9bd1190609c21902f23dc840135dca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 03:55:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f13325a5d6640ccad57867693cf8ce9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ed9cf5220fb4d16b3438a14d22fac05 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e36d2d76a9d485286a24c21d3982fea : SUCCESS in 3m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ad625664609c4bd29f414f576a6e1046 : SUCCESS in 2m 52s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/57ee23e87dd947eeb3cebb1ef4c10982 : FAILURE in 4m 05s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1e391ddf240a4a9ca59e47f59fde7b95 : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c7e7e2e6f29ada9aefab2b6d054a1eeb46173f32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35271,"name":"Shuai Qian","display_name":"qian.shuai","email":"qs_cloudnative@163.com","username":"iauhsnaiq"},"date":"2023-02-15 05:38:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c94ea86d453268f9d8e578e33544b381e83317fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 05:51:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf5f047b4c6c4df19c83f8d64b09ee7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a843bf5cc04830974e5ef8fbe60a86 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/759be5f3076d436cb9d6a811f1ff16b7 : SUCCESS in 2m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/350706b760644f5d854dc1d0c83250d3 : SUCCESS in 2m 44s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/9c32e9f7795f46d2a67aae18fccf051f : FAILURE in 3m 23s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/4c8d4c0d05d54363ab081357f1097d88 : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ef1675b3ffdd5f631a7b0096a78f542d7d1ff2b8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35271,"name":"Shuai Qian","display_name":"qian.shuai","email":"qs_cloudnative@163.com","username":"iauhsnaiq"},"date":"2023-02-15 06:25:41.000000000","message":"Abandoned\n\nI Do Not Deserve it, 555.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2e1248df201971fc7547e22866fb6dd259d9b05","revisions":{"11b5df8e0497b91a8ac5f3155e60e1264c3b90e0":{"kind":"REWORK","_number":1,"created":"2023-02-15 03:45:20.000000000","uploader":{"_account_id":35271,"name":"Shuai Qian","display_name":"qian.shuai","email":"qs_cloudnative@163.com","username":"iauhsnaiq"},"ref":"refs/changes/51/873851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/51/873851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/51/873851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/51/873851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/51/873851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/51/873851/1"}}},"commit":{"parents":[{"commit":"6a27cb1846f900a7f7cc4399a44ee260d47f9b35","subject":"Merge \"Add Slawek Kaplonski for TC in Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6a27cb1846f900a7f7cc4399a44ee260d47f9b35"}]}],"author":{"name":"iauhsnaiq","email":"qs_cloudnative@163.com","date":"2023-02-15 03:43:31.000000000","tz":480},"committer":{"name":"iauhsnaiq","email":"qs_cloudnative@163.com","date":"2023-02-15 03:45:08.000000000","tz":480},"subject":"Add qian.shuai candidacy for Magnum PTL","message":"Add qian.shuai candidacy for Magnum PTL\n\nChange-Id: Ifbd5e3ac0a18e0492ea2b700d15462eaab676b02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/11b5df8e0497b91a8ac5f3155e60e1264c3b90e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/11b5df8e0497b91a8ac5f3155e60e1264c3b90e0"}]},"branch":"refs/heads/master"},"b2e1248df201971fc7547e22866fb6dd259d9b05":{"kind":"REWORK","_number":2,"created":"2023-02-15 05:38:40.000000000","uploader":{"_account_id":35271,"name":"Shuai Qian","display_name":"qian.shuai","email":"qs_cloudnative@163.com","username":"iauhsnaiq"},"ref":"refs/changes/51/873851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/51/873851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/51/873851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/51/873851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/51/873851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/51/873851/2"}}},"commit":{"parents":[{"commit":"6a27cb1846f900a7f7cc4399a44ee260d47f9b35","subject":"Merge \"Add Slawek Kaplonski for TC in Bobcat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6a27cb1846f900a7f7cc4399a44ee260d47f9b35"}]}],"author":{"name":"iauhsnaiq","email":"qs_cloudnative@163.com","date":"2023-02-15 03:43:31.000000000","tz":480},"committer":{"name":"iauhsnaiq","email":"qs_cloudnative@163.com","date":"2023-02-15 05:38:30.000000000","tz":480},"subject":"Add qian.shuai candidacy for Magnum PTL","message":"Add qian.shuai candidacy for Magnum PTL\n\nChange-Id: Ifbd5e3ac0a18e0492ea2b700d15462eaab676b02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b2e1248df201971fc7547e22866fb6dd259d9b05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b2e1248df201971fc7547e22866fb6dd259d9b05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
