)]}'
{"id":"openstack%2Felection~874970","triplet_id":"openstack%2Felection~master~I628cd209dbb2ce6217a8738208c6f8910452e96d","project":"openstack/election","branch":"master","topic":"b-election","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2023-02-24 01:28:58.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-02-24 01:17:53.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"I628cd209dbb2ce6217a8738208c6f8910452e96d","subject":"Bobcat Election Results (TC/PTL)","status":"MERGED","created":"2023-02-23 23:19:31.000000000","updated":"2023-02-24 01:29:53.000000000","submitted":"2023-02-24 01:28:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874970-b-election","meta_rev_id":"bdddc214cd822adfddebd6a7702e90aa9f61be5a","_number":874970,"virtual_id_number":874970,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-24 01:28:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2023-02-24 01:17:53.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2023-02-23 23:24:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2023-02-24 01:17:53.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 23:24:23.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-02-23 23:25:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 00:14:23.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"4b4186547865d9358d48e8aebf1ac4662ea7dbd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-02-23 23:19:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaab9833bd521f1cbca43660ad846c9642a4091f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-23 23:24:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58ce85875f2dfc907a94c64c934ec03242cecaca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 23:25:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc6aa546f05745fdbf8dd5e022501dda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa50ea24aabd464ebeade620c3b51f11 : SUCCESS in 4m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c5d49fb4528a4e728850e56455c4341b : SUCCESS in 2m 46s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8e184756bd73467e940bf690738e1054 : SUCCESS in 2m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/50f84a678d444e7cab5f652393102c4d : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3cbf4f31fbc655fbb0410e35600077315609b1f6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-02-24 01:17:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de6494e764e4d50dabd8267c1a0fc947193df11c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 01:18:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5d3c549a0912b4c38382a12a8cf4f256e2c2d48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 01:28:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfed6ecf57c748ef8fec55f3a06552d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9cf26889dfd46c7ad14c7f9d333ecc6 : SUCCESS in 4m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c84fddb54fa647e3aed1a9bcc853960a : SUCCESS in 3m 16s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b0481434a7c14496bd038dd442f1ed8d : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7a2e1ada3d1e7ac2a2734392d77a151a6001c484","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 01:28:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bdddc214cd822adfddebd6a7702e90aa9f61be5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 01:29:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/502b3d28be124633980c1f97b677eb81\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/57913adccddd4b66a5df9df673795237 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"349b659a038993b241b17b0514516d49f3d2c592","revisions":{"349b659a038993b241b17b0514516d49f3d2c592":{"kind":"REWORK","_number":1,"created":"2023-02-23 23:19:31.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/70/874970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/70/874970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/70/874970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/70/874970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/70/874970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/70/874970/1"}}},"commit":{"parents":[{"commit":"80f2ccfd426bf4bbd1a90c9e4758fbbb7dbeaef3","subject":"Adding Dave Wilde candidacy for the Keystone project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/80f2ccfd426bf4bbd1a90c9e4758fbbb7dbeaef3"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-02-23 23:12:00.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-02-23 23:12:00.000000000","tz":540},"subject":"Bobcat Election Results (TC/PTL)","message":"Bobcat Election Results (TC/PTL)\n\nClose out Bobcat / 2023.2 elections.\n\nChange-Id: I628cd209dbb2ce6217a8738208c6f8910452e96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/349b659a038993b241b17b0514516d49f3d2c592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/349b659a038993b241b17b0514516d49f3d2c592"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
