)]}'
{"id":"openstack%2Felection~875268","triplet_id":"openstack%2Felection~master~Ic35015161628e8a24976d0d2955b828d460f7863","project":"openstack/election","branch":"master","topic":"migrate-release-version","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-03-04 07:18:49.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-03-04 07:18:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic35015161628e8a24976d0d2955b828d460f7863","subject":"DNM: testing create dir: migration election tooling to release version","status":"ABANDONED","created":"2023-02-25 02:49:41.000000000","updated":"2023-03-04 07:18:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"81cb043e9fa4a9abc4d98994e0fd55b90b9c66fc","_number":875268,"virtual_id_number":875268,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-03 20:58:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-25 02:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-02 01:27:46.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"efa7c15ba4ce2a2b0959c477abbb8e6ae4697c19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-25 02:49:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4d8e21b2bf0af9fdf5c0a2227ab439fdb78ce55","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-25 02:53:03.000000000","message":"Topic testing removed","accounts_in_message":[],"_revision_number":1},{"id":"c3156e3f1463873c731ac35f5d871b607022bc68","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-25 02:53:06.000000000","message":"Topic set to migrate-release-version","accounts_in_message":[],"_revision_number":1},{"id":"01559630ec48b5fd63da9521b440a91c10cad668","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-25 02:56:30.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2195c86ccd1e6f29ba18ce9038bcb4b998a6b931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 02:59:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef0e23bbc24c44a286bdb68d72c9a836\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e13c9a29c0740129600145ba1defcd0 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05ab6a7fa11d45579675bcd8d0058206 : SUCCESS in 2m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3dcf6adce97b49068d69aeb64129174f : FAILURE in 3m 45s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/5ed0265ad02c41f3a463fe4160236633 : SUCCESS in 3m 28s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5f0960696cc44159831d49c51f282b26 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"be7424a05f913f54e9bf430425f00759920f4c3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-25 03:13:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95cbad5298f1389be66487cd1aab0cf714ec2a0f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-25 03:15:29.000000000","message":"Topic testing4 removed","accounts_in_message":[],"_revision_number":2},{"id":"6f7a4d5cb9933d501ce94f27752d4e83e66ecc4d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-25 03:15:32.000000000","message":"Topic set to migrate-release-version","accounts_in_message":[],"_revision_number":2},{"id":"2888000cf8283ec42eb0fc9719ce97ce42cc6447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 03:18:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0df68bba31164228a0d75711141f15f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b0fa19f6c3041a9af814c07d46192bd : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1be47157b85947d7a37869f88974ed50 : SUCCESS in 2m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2599482ce9f4d50a7f7fe94d80c435f : SUCCESS in 3m 03s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2ca36d56e32c402b9417fcca20db733a : SUCCESS in 2m 48s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/11d94e5809684a449681323a7db23e45 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"7ab070693fee1c232226ca857ead6233c55aaf8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-03-02 01:27:46.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6f2058179a26fa05ff88be9d1ca72dcfc808c9eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 01:39:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a980e09157c04faaa038bd8b1844afe2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf39f99feeb462fbc6a84ff2754ec73 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbdc6511527d47a8809f737456b040e9 : SUCCESS in 2m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fc9ce25cae704dec8e2e2d7f04a47301 : FAILURE in 2m 42s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/cbccf0d6142b47fc945b4572fa4141df : SUCCESS in 3m 03s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c52bd74be7c24d2c9c6e186f6179486c : SUCCESS in 2m 34s\n\nWarning:\n  Comments left for invalid file openstack_election/cmds/setup_election_config.py","accounts_in_message":[],"_revision_number":3},{"id":"11c4f85c61e67adc57cd40350951b9ef272f0554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-03 05:00:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d0c54bd01a3bec1971b047295230e5381ff0c50e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 05:06:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c62a9d6ea7c4637a7ddebf44c05427c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0dab19b0d304c3fb71c5a2e1c82bfe1 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/322a8ac43d2649a081e6caf1a5868b3d : SUCCESS in 3m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e76f7467b3ef45a08fc2e59026443b63 : SUCCESS in 3m 31s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8d34d5670a7d459b99b497e38bd8144f : SUCCESS in 3m 15s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/25d20bed948a48de9464e8ab5057bbeb : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"928494cbad64a00e821f0840bc297618148fbc41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-03 19:05:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"73672c1d4663526b43ccc7c00001e557ba48ef84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 19:16:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc17241a65734f0a9011483ca2824b73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22252fc47bd64700a6bce1c825bcd67d : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fddb3c2128740bb9b12987687a62d47 : SUCCESS in 2m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/88b9a1cbff2a42d2a023dc42e8bef0c2 : FAILURE in 3m 26s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/bbb253649d68406bb60b125c1058583e : SUCCESS in 3m 35s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c11ad68eb38743faa1411bc88ba07663 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":5},{"id":"e502c0d33644ca2d4832b3d03fd0caf32ab20e3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-03 20:37:49.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"4de655b5ef8bb30a190aa342b7935140d23e6438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 20:58:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21af517321dc4724994b06dff9cced4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aec9c062edef401f97a680cd74d12a3e : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/450e7abc92894cd3b678bd86ec0b8533 : SUCCESS in 2m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0e6caf0ac2004fcaa8823d4f76e6c31f : SUCCESS in 2m 33s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/65d402fa377c4b0ca0745db85eafe927 : SUCCESS in 2m 58s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ec5a93904d22406983c5eff0eb819294 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":6},{"id":"81cb043e9fa4a9abc4d98994e0fd55b90b9c66fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-03-04 07:18:49.000000000","message":"Abandoned\n\nNo Longer needed","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"567446458a298d7921f61aac3286d1e1780ba50d","revisions":{"09de48454fc92abe49d70a0cd3716a90a65e11e3":{"kind":"REWORK","_number":1,"created":"2023-02-25 02:49:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/875268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/875268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/875268/1"}}},"commit":{"parents":[{"commit":"f3ef0a78524d62431e9d9f78933359cba9fd8cc0","subject":"Migrate election tooling to release versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f3ef0a78524d62431e9d9f78933359cba9fd8cc0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-25 02:49:15.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-25 02:49:15.000000000","tz":-480},"subject":"DNM: testing release version migration change","message":"DNM: testing release version migration change\n\nChange-Id: Ic35015161628e8a24976d0d2955b828d460f7863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/09de48454fc92abe49d70a0cd3716a90a65e11e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/09de48454fc92abe49d70a0cd3716a90a65e11e3"}]},"branch":"refs/heads/master"},"16a60382d533932b0eecdc893377c3dbaf6803e0":{"kind":"REWORK","_number":2,"created":"2023-02-25 03:13:03.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/875268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/875268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/875268/2"}}},"commit":{"parents":[{"commit":"dc8a387283808fa328e927d5036270ec52c89933","subject":"DNM: testing configuration: migration election tooling to release version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dc8a387283808fa328e927d5036270ec52c89933"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-25 03:12:36.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-25 03:12:36.000000000","tz":-480},"subject":"DNM: testing create dir: migration election tooling to release version","message":"DNM: testing create dir: migration election tooling to release version\n\nChange-Id: Ic35015161628e8a24976d0d2955b828d460f7863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/16a60382d533932b0eecdc893377c3dbaf6803e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/16a60382d533932b0eecdc893377c3dbaf6803e0"}]},"branch":"refs/heads/master"},"c93a4a4b05f7b953618664157ba2f707d41a727b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-02 01:27:46.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/68/875268/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/875268/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/875268/3"}}},"commit":{"parents":[{"commit":"141c289038f359915c5dd57eaf619e1b9dfa4ba0","subject":"Squash into Ia8a60a9514b1dca5ac97ed8099c34c6c6c7d24e2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/141c289038f359915c5dd57eaf619e1b9dfa4ba0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-25 03:12:36.000000000","tz":-480},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-03-02 01:13:18.000000000","tz":660},"subject":"DNM: testing create dir: migration election tooling to release version","message":"DNM: testing create dir: migration election tooling to release version\n\nChange-Id: Ic35015161628e8a24976d0d2955b828d460f7863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c93a4a4b05f7b953618664157ba2f707d41a727b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c93a4a4b05f7b953618664157ba2f707d41a727b"}]},"branch":"refs/heads/master"},"e8ba8718ce250864cb89b8841c4e736a45a6d4d9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-03-03 05:00:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/875268/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/875268/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/875268/4"}}},"commit":{"parents":[{"commit":"ecd5497dc1979c5a227fe5433fc5b1483e060ccb","subject":"DNM: testing configuration: migration election tooling to release version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ecd5497dc1979c5a227fe5433fc5b1483e060ccb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-25 03:12:36.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-03-03 05:00:36.000000000","tz":0},"subject":"DNM: testing create dir: migration election tooling to release version","message":"DNM: testing create dir: migration election tooling to release version\n\nChange-Id: Ic35015161628e8a24976d0d2955b828d460f7863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e8ba8718ce250864cb89b8841c4e736a45a6d4d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e8ba8718ce250864cb89b8841c4e736a45a6d4d9"}]},"branch":"refs/heads/master","description":"Rebase"},"88fc7facfcc38af7799bd3a480e02a5c7d895ddc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-03 19:05:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/875268/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/875268/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/875268/5"}}},"commit":{"parents":[{"commit":"87ff48bdecd7aba80ecf630f1dadb705e3eaedc6","subject":"DNM: testing configuration: migration election tooling to release version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/87ff48bdecd7aba80ecf630f1dadb705e3eaedc6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-25 03:12:36.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-03-03 19:05:28.000000000","tz":0},"subject":"DNM: testing create dir: migration election tooling to release version","message":"DNM: testing create dir: migration election tooling to release version\n\nChange-Id: Ic35015161628e8a24976d0d2955b828d460f7863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/88fc7facfcc38af7799bd3a480e02a5c7d895ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/88fc7facfcc38af7799bd3a480e02a5c7d895ddc"}]},"branch":"refs/heads/master","description":"Rebase"},"567446458a298d7921f61aac3286d1e1780ba50d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-03-03 20:37:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/875268/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/875268/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/875268/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/875268/6"}}},"commit":{"parents":[{"commit":"6fba731c3d8ef71d54c075ecf8f13f780cd2ff2e","subject":"DNM: testing configuration: migration election tooling to release version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/6fba731c3d8ef71d54c075ecf8f13f780cd2ff2e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-02-25 03:12:36.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-03-03 20:37:49.000000000","tz":0},"subject":"DNM: testing create dir: migration election tooling to release version","message":"DNM: testing create dir: migration election tooling to release version\n\nChange-Id: Ic35015161628e8a24976d0d2955b828d460f7863\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/567446458a298d7921f61aac3286d1e1780ba50d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/567446458a298d7921f61aac3286d1e1780ba50d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
