)]}'
{"id":"openstack%2Felection~875998","triplet_id":"openstack%2Felection~master~I61c472076520b15c90fc27e305b1a336aebd460d","project":"openstack/election","branch":"master","topic":"migrate-release-version","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-03-03 07:22:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I61c472076520b15c90fc27e305b1a336aebd460d","subject":"Squash into Ia8a60a9514b1dca5ac97ed8099c34c6c6c7d24e2","status":"ABANDONED","created":"2023-03-02 01:27:46.000000000","updated":"2023-03-03 07:22:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2d40eb05b6c394c4dff9f99a7e62c9ebe90fc527","_number":875998,"virtual_id_number":875998,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-02 01:37:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 01:37:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a65b5df01e07541786c42201f5c3c79b1e49e151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-03-02 01:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24e6ef63d8f032a04348e5348f68c6bc2c348c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 01:37:58.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41cc17f4c40146b3904dd025e29f22cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ccdde31b0364f6b8ec635384bdf0575 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/171e8fbb2c5147049a8020375e7191ee : SUCCESS in 2m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/070c5f2d182c4936a1802cf8aea20918 : FAILURE in 2m 39s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8bb01d2ee3974faa8d2e6502c501fce2 : FAILURE in 2m 33s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/50b9feefa58345d19401687c61bba08d : FAILURE in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2d40eb05b6c394c4dff9f99a7e62c9ebe90fc527","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-03-03 07:22:58.000000000","message":"Abandoned\n\nNo longer needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"141c289038f359915c5dd57eaf619e1b9dfa4ba0","revisions":{"141c289038f359915c5dd57eaf619e1b9dfa4ba0":{"kind":"REWORK","_number":1,"created":"2023-03-02 01:27:46.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/98/875998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/98/875998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/98/875998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/98/875998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/98/875998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/98/875998/1"}}},"commit":{"parents":[{"commit":"d12ca8716ccfe8b531c496c7dd05e61f27b8436b","subject":"DNM: testing configuration: migration election tooling to release version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d12ca8716ccfe8b531c496c7dd05e61f27b8436b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-03-02 00:44:02.000000000","tz":660},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-03-02 01:11:58.000000000","tz":660},"subject":"Squash into Ia8a60a9514b1dca5ac97ed8099c34c6c6c7d24e2","message":"Squash into Ia8a60a9514b1dca5ac97ed8099c34c6c6c7d24e2\n\nEven though we want to refer to a release with it\u0027s semver/datever\nlike identifier which can be treated like a float, most of the code\nactaully uses it like a string.  As such force the version to be a\nstring.  We could do this in the yaml itself with things like\n\"release: !!str 2023.2\";  or release: \u00272023.2\u0027 but lets do it in the\ncode because it\u0027s easier?\n\nChange-Id: I61c472076520b15c90fc27e305b1a336aebd460d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/141c289038f359915c5dd57eaf619e1b9dfa4ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/141c289038f359915c5dd57eaf619e1b9dfa4ba0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
