)]}'
{"id":"openstack%2Felection~876738","triplet_id":"openstack%2Felection~master~I7275411fad3887cc6b9d5bf9152231a6cc6fddba","project":"openstack/election","branch":"master","attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-05-13 17:18:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15993\u003e replied on the change","reason_account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7275411fad3887cc6b9d5bf9152231a6cc6fddba","subject":"Add affiliation lookup functionality","status":"NEW","created":"2023-03-07 14:13:15.000000000","updated":"2025-10-15 01:41:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"13c5bacf0170333f3fa153fc7ec5f991c41943a4","_number":876738,"virtual_id_number":876738,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-15 01:41:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}],"REVIEWER":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}],"REVIEWER":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-03-07 14:22:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-31 16:10:57.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2025-05-13 17:18:53.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"}],"messages":[{"id":"d4a83697d8802d6bc0385871bcf646fb10e3cf74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-03-07 14:13:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18522484b93cf22b2906f556e783c4e899d43eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 14:22:05.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/628cff5b8c234c99b8476f585b50c632\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e43eac4b004971aa96c25a854b2b53 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95d65613bbb448f88cd60f73674a6cf2 : SUCCESS in 2m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3962628838f34120bd7848822de98226 : FAILURE in 2m 41s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/5266f2c2bfed436794fca488e6643fcf : SUCCESS in 3m 08s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/21d15893588548ada3f4db7db10b9034 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f015df1ba4212a7bc38ee58638108757f29dcfa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-03-07 14:25:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"75075ec7f59ee43d1948556a8b39bb853fabd13a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-03-07 14:27:05.000000000","message":"Set Work In Progress\n\nThis still just dumps YAML to stdout, and needs some configuraiton options for consistency with other tools in the repo.","accounts_in_message":[],"_revision_number":2},{"id":"9e9ad2b63c16a6f948b596313a35c24bf35ccb9f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-03-07 14:27:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ff8fd396df57bf25e1b4c7a9753f2fdad1cd2ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 14:56:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96cf1239b06b49b8af49c9d6e10c96d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83474f3b82c848548dc385c2ba29d0b2 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65d09036de9945d8b0a78f5d78aab4fa : SUCCESS in 2m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/05e4aa1d7f484431a7cb25576fa1631d : SUCCESS in 2m 31s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/06d091569eeb478291bcd0a44837d1f4 : SUCCESS in 2m 32s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b4f45774546b4b2abc2e66f15248fc5f : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"93a8a691953279a3aee6c076cb3b3b2b5fe4beca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-12-06 21:40:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bedde053ffe691e06091ab2be01fd6a72d37ad7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 21:50:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7512c1a4bccf4d38bfe5933522649407\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b64d867bc134003a1fde75c92597090 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14a51f265d23401a921a767cbc1b29f7 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fac69bf9f794c3eb0a03feaa47dc011 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4fd4a5c40a584bfe993fea31295d3885 : SUCCESS in 3m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f0976998bffb4944ac7d97fc3fc50c3e : SUCCESS in 3m 02s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/f29d3380934e4e0d982d6971c57cc5ab : SUCCESS in 3m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/94dbd3abaf744899b9598e86a49dfa5c : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0bd6600c2c0b1b14d53f36880bb913073526db58","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-01-26 18:45:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c9e5b64deab8381214e6380e15ef6adc85323e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 19:24:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b430d6fe8ad496b87b89a50bb88bb70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50caa36fff3a4330be51b627d1238c1d : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08be9831d85f498f938dfdfa78bb3aae : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17a35dc232a14aa39409ce855fa8d383 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19399afa2af3475ba54a289bc5e2a5b9 : SUCCESS in 4m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0d7ab566796a483592f5dad47b6d4e03 : SUCCESS in 2m 52s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6e8c9247a1f04967a0c9664ad5e944e0 : SUCCESS in 3m 20s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/21c7816c30254eeda3d1336c8d766fc7 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":4},{"id":"33debb71e3dc3216facfc888c2bbdc0a31893965","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-07-31 16:10:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3d9f34f345ebd08bd2d7926db1b539a311b8fc0d","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-07-31 16:11:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3cafe209de15ccf99fcabfbe9383b4c30a6e5aca","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-07-31 16:19:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f91cad4a48dcff9f5bd22e04d30804bfc6919fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 16:19:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/863ea3b4a3d54b5fa3e2a19dc6be841f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8b79a2835d4f3bb66cdccf8704bc52 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dc1ef106e1e471e8b6aa755034b7b21 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1b1e1d0f76c4deca23eaf88dd897100 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6986363a799e427ba357d78793417e78 : SUCCESS in 3m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d4a91a8c4a494eec8743f3ccdb7f4a78 : SUCCESS in 2m 41s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/82caa515ef3645029bee8d9845722d6f : SUCCESS in 2m 30s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/294cf088558d4db498429183ac2d708b : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":4},{"id":"3b2d81b959107d5bbdb1a819a7621329dbab0e25","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-07-31 23:55:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"de1c22237f01e89c313f736bc631220ec2e80d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 23:59:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a86188f88d64d009d2ef0dc25094f79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0948525d28514430ae2cecfae2e6087d : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffe88d25815b441eb9a3324acf91199f : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20271502d6d54a729ad052e8dec5dcb4 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23d07a1196e94c0ea3754ae7b80c2e96 : SUCCESS in 2m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bc06e09369c440faa2ad157132a5e52f : SUCCESS in 3m 03s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/1ddd7319f5c44aee97f5c08e9596bce0 : SUCCESS in 2m 56s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ace6e38314d34a96ba6b850ebe6992ed : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":5},{"id":"ccc559a78fc6eed676fb0b1294ec00f7b49facfd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-08-06 17:55:34.000000000","message":"Patch Set 5:\n\nNote that the addition to openstack_election/owners.py could be split out to its own change, as it\u0027s more generally useful beyond just post-election leadership affiliation tracking.","accounts_in_message":[],"_revision_number":5},{"id":"b0560b0174941a3a4b6a64ebe60a7433b9dd9ef8","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2025-05-13 17:18:53.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"5a2814233b9bf49278a193a5a62da4b862eae73c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-10-15 00:53:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fc5b62a444054f2592144ae64a013ea0450ab01c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 01:02:53.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d45011d478924615beb5aa9c47e778ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96b27f5ac9b4ad0b09bb28efc3d4401 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d29df0f7726a48e99de9e89e1edee1fc : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce79c0bbe20f46f7bc78dcaced9aa1e3 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7b3e5800d0d436d9112e3af0e2a9687 : SUCCESS in 3m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6a918ed03c604bc18fe0971970a19fde : FAILURE in 3m 55s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e420ffe05ba4413b9c951c77695ec976 : SUCCESS in 4m 50s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/20195d3b13e34447a8ba32a554002a03 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":6},{"id":"06570df9d168ed95e12489c467ebb82aac9682d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-10-15 01:33:06.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"13c5bacf0170333f3fa153fc7ec5f991c41943a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 01:41:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5670accb1efa44f29740a93ed3adad37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c709d492dee404ea236268bc4be1842 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f99b139a262470eb219432fe081343a : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d539f74c2ad4762960b4dc57beac884 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b4240a8578a484eab0fefc12cf49c91 : SUCCESS in 3m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3b67cd148f87456d83ae507002bdf7c9 : SUCCESS in 3m 46s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/35db58394ecc4e6c95129e60b0a56c7e : SUCCESS in 3m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/77cb206e792f49dda8bf2bb0efd695ef : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1c02a1bb18eeb070608b8b5831ebf51c54c7bedc","revisions":{"da1069b92cfc7a40d0ee35d197f1ba79dcaf2840":{"kind":"REWORK","_number":1,"created":"2023-03-07 14:13:15.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/38/876738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/876738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/876738/1"}}},"commit":{"parents":[{"commit":"349b659a038993b241b17b0514516d49f3d2c592","subject":"Bobcat Election Results (TC/PTL)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/349b659a038993b241b17b0514516d49f3d2c592"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-03-07 14:09:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-03-07 14:09:54.000000000","tz":0},"subject":"Add affiliation lookup functionality","message":"Add affiliation lookup functionality\n\nPer OpenInfra Foundation bylaws, fewer than half of the TC members\nare allowed t be affiliated with the same organization. In order to\nmake it easier to find this information, add scripts to look up\naffiliation information in the foundation member directory.\n\nChange-Id: I7275411fad3887cc6b9d5bf9152231a6cc6fddba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/da1069b92cfc7a40d0ee35d197f1ba79dcaf2840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/da1069b92cfc7a40d0ee35d197f1ba79dcaf2840"}]},"branch":"refs/heads/master"},"d08bd815bd9cae6b3b1329d8e7611ad927b7c6d2":{"kind":"REWORK","_number":2,"created":"2023-03-07 14:25:23.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/38/876738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/876738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/876738/2"}}},"commit":{"parents":[{"commit":"349b659a038993b241b17b0514516d49f3d2c592","subject":"Bobcat Election Results (TC/PTL)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/349b659a038993b241b17b0514516d49f3d2c592"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-03-07 14:09:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-03-07 14:25:08.000000000","tz":0},"subject":"Add affiliation lookup functionality","message":"Add affiliation lookup functionality\n\nPer OpenInfra Foundation bylaws, fewer than half of the TC members\nare allowed t be affiliated with the same organization. In order to\nmake it easier to find this information, add scripts to look up\naffiliation information in the foundation member directory.\n\nChange-Id: I7275411fad3887cc6b9d5bf9152231a6cc6fddba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d08bd815bd9cae6b3b1329d8e7611ad927b7c6d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d08bd815bd9cae6b3b1329d8e7611ad927b7c6d2"}]},"branch":"refs/heads/master"},"21cd3533ddfa587d16bbceed498c48029ab91fa9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-06 21:40:54.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/38/876738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/876738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/876738/3"}}},"commit":{"parents":[{"commit":"39a160dc931ca66a84c0e3bc6a1fc88f8fb91b7d","subject":"Merge \"[tox] Exclude the build directory from flake8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/39a160dc931ca66a84c0e3bc6a1fc88f8fb91b7d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-03-07 14:09:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-12-06 17:01:24.000000000","tz":0},"subject":"Add affiliation lookup functionality","message":"Add affiliation lookup functionality\n\nPer OpenInfra Foundation bylaws, fewer than half of the TC members\nare allowed t be affiliated with the same organization. In order to\nmake it easier to find this information, add scripts to look up\naffiliation information in the foundation member directory.\n\nChange-Id: I7275411fad3887cc6b9d5bf9152231a6cc6fddba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/21cd3533ddfa587d16bbceed498c48029ab91fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/21cd3533ddfa587d16bbceed498c48029ab91fa9"}]},"branch":"refs/heads/master"},"7628eb8f88c41c49899e98e05b30ac5a9e095766":{"kind":"REWORK","_number":4,"created":"2024-01-26 18:45:09.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/38/876738/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/876738/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/876738/4"}}},"commit":{"parents":[{"commit":"15a6518a783cbd4df29cd76a5b553a42fd9caae9","subject":"Get historical affiliations from the summit API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/15a6518a783cbd4df29cd76a5b553a42fd9caae9"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-03-07 14:09:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-01-26 18:45:00.000000000","tz":0},"subject":"Add affiliation lookup functionality","message":"Add affiliation lookup functionality\n\nPer OpenInfra Foundation bylaws, fewer than half of the TC members\nare allowed t be affiliated with the same organization. In order to\nmake it easier to find this information, add scripts to look up\naffiliation information in the foundation member directory.\n\nChange-Id: I7275411fad3887cc6b9d5bf9152231a6cc6fddba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7628eb8f88c41c49899e98e05b30ac5a9e095766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7628eb8f88c41c49899e98e05b30ac5a9e095766"}]},"branch":"refs/heads/master"},"9e1b1472224d331dcc1e747c01ea920f2c357a02":{"kind":"REWORK","_number":5,"created":"2024-07-31 23:55:41.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/38/876738/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/876738/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/876738/5"}}},"commit":{"parents":[{"commit":"77767954a13bc813ac3b3b70e03f4f2f04ca122a","subject":"Merge \"Update PTL and TC election dates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/77767954a13bc813ac3b3b70e03f4f2f04ca122a"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-03-07 14:09:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-07-31 23:55:30.000000000","tz":0},"subject":"Add affiliation lookup functionality","message":"Add affiliation lookup functionality\n\nPer OpenInfra Foundation bylaws, fewer than half of the TC members\nare allowed t be affiliated with the same organization. In order to\nmake it easier to find this information, add scripts to look up\naffiliation information in the foundation member directory.\n\nChange-Id: I7275411fad3887cc6b9d5bf9152231a6cc6fddba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9e1b1472224d331dcc1e747c01ea920f2c357a02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9e1b1472224d331dcc1e747c01ea920f2c357a02"}]},"branch":"refs/heads/master"},"dfe07318dcd123103b4b5f95eb14f080ebf0ea16":{"kind":"REWORK","_number":6,"created":"2025-10-15 00:53:20.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/38/876738/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/876738/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/876738/6"}}},"commit":{"parents":[{"commit":"77767954a13bc813ac3b3b70e03f4f2f04ca122a","subject":"Merge \"Update PTL and TC election dates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/77767954a13bc813ac3b3b70e03f4f2f04ca122a"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-03-07 14:09:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-10-15 00:53:08.000000000","tz":0},"subject":"Add affiliation lookup functionality","message":"Add affiliation lookup functionality\n\nPer OpenInfra Foundation bylaws, fewer than half of the TC members\nare allowed t be affiliated with the same organization. In order to\nmake it easier to find this information, add scripts to look up\naffiliation information in the foundation member directory.\n\nChange-Id: I7275411fad3887cc6b9d5bf9152231a6cc6fddba\nSigned-off-by: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dfe07318dcd123103b4b5f95eb14f080ebf0ea16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dfe07318dcd123103b4b5f95eb14f080ebf0ea16"}]},"branch":"refs/heads/master"},"1c02a1bb18eeb070608b8b5831ebf51c54c7bedc":{"kind":"REWORK","_number":7,"created":"2025-10-15 01:33:06.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/38/876738/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/876738/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/876738/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/876738/7"}}},"commit":{"parents":[{"commit":"1d0138f76bd334288ee8f79685db2c8202b4e358","subject":"Merge \"Update tc-election-summary.py: reflect 2026.1 results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1d0138f76bd334288ee8f79685db2c8202b4e358"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2023-03-07 14:09:54.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-10-15 01:33:05.000000000","tz":0},"subject":"Add affiliation lookup functionality","message":"Add affiliation lookup functionality\n\nPer OpenInfra Foundation bylaws, fewer than half of the TC members\nare allowed t be affiliated with the same organization. In order to\nmake it easier to find this information, add scripts to look up\naffiliation information in the foundation member directory.\n\nChange-Id: I7275411fad3887cc6b9d5bf9152231a6cc6fddba\nSigned-off-by: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1c02a1bb18eeb070608b8b5831ebf51c54c7bedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1c02a1bb18eeb070608b8b5831ebf51c54c7bedc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
