)]}'
{"id":"openstack%2Felection~881138","triplet_id":"openstack%2Felection~master~I1ea1e98ba4b7214b4576ecac1e2fdd66bd218c06","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-05-16 02:08:53.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-05-16 02:52:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ea1e98ba4b7214b4576ecac1e2fdd66bd218c06","subject":"Switch to 2023.2 testing runtime py version","status":"MERGED","created":"2023-04-20 19:25:48.000000000","updated":"2023-05-16 02:54:52.000000000","submitted":"2023-05-16 02:52:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881138","meta_rev_id":"3bd3237b3ac292c64f24f6a2b6ad1684334e6650","_number":881138,"virtual_id_number":881138,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-16 02:52:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2023-05-16 02:08:53.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2023-05-15 18:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2023-05-16 02:08:53.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 19:44:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 09:33:10.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2023-05-15 18:13:14.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"6fd6648e421cabe22d35cc0105d17d0187b072ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-20 19:25:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"301b883b4a9368c6c38a2a0e71d8bed077209dc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:44:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75817ee13f8b4874bec66136981515cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af0af979cb346c89ec5c611a7367422 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7c9fc336ef4be5aa9f85904fce89e7 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07a18a22cbaa442dab2f5ce326382b9a : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9af9eda600d45a5bd4b1217da5f6d65 : SUCCESS in 3m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6fe0c5a16201404cb7f8f0270bbd7590 : SUCCESS in 3m 11s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e7bb1fe355384a04a26bf3ddc89f1585 : SUCCESS in 3m 01s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/30e5c429ffe04b0f80a5d940ac662921 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"35be20468ebef36fb37a583aa211a87e2aa90074","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-05-10 09:33:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f9c9a8c1f47db26212613d75cff27852f54f031","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-10 15:29:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80c3c20bd7a393cb1f652e7b5714946dd6f69826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-10 15:29:54.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fc06812133f1a20f7a8a913982a356120f656a19","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-10 15:30:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7024fe14dbf341386763d80066a8b01bfcce191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 15:41:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b53e31339954b1bb71adaececdf7f92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f626e096fa61444398918c13112aed1a : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdaa0f40b4264e79a80803e90bad05a1 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba33601c9d8745cd9069ec6c5a8f70f9 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d173e391408d403f873ac202c4677fef : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ddafc9e2c85460da85a7ebf62dd5e19 : SUCCESS in 4m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2a5bff57df144fd799948f9b3866e637 : SUCCESS in 3m 10s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a45f18f6eb30449ca43d6f3cc193d09e : SUCCESS in 2m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/45343b216fda4234bd108b1f278990ad : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"2090943010addce906ec6640d58999f68aaccfeb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-10 16:51:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd3b35cf5c86c69ddda2a2a02ad5bdd1683332fd","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-05-15 18:13:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0408c5798cfd37a0e7f11e24b073bb83f69edd4b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-05-16 02:08:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ceb797b418f8af92d0fdbc7f54e9fb4b344bafa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 02:11:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6359cc43851fb42860056dc7afae3ac0bddab45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 02:52:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43dae13989e247209cda345001d246d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb3e236824c4342bd657467e446ee79 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa607b190a4740c289e36b7e3df4587f : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a309ad56bb7d4c2a86277c280f2d9efc : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b04fbb88bc2e46389f8f0588946bca88 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d42dd72c61646ea997a50e96c349cfd : SUCCESS in 3m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f2c7b199e97442bd952e24f893fd6870 : SUCCESS in 2m 59s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7ee8d0fa453c459a81efb3f7b9289994 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b2a87a62a0eb3fcdbb138ef5cd8a02e820e27a44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 02:52:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3bd3237b3ac292c64f24f6a2b6ad1684334e6650","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 02:54:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7175efa7c21437f9a23c0a63fed4910\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/773e223f31194caeb42699be22865b53 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f56f206e623fc6d984f4737d5c85bb38beb80cc","revisions":{"7854ed71c7b13a7ef6c6d6e7b13522e9bd1ab63a":{"kind":"REWORK","_number":1,"created":"2023-04-20 19:25:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/38/881138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/881138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/881138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/881138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/881138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/881138/1"}}},"commit":{"parents":[{"commit":"349b659a038993b241b17b0514516d49f3d2c592","subject":"Bobcat Election Results (TC/PTL)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/349b659a038993b241b17b0514516d49f3d2c592"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-04-20 19:25:09.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-04-20 19:25:09.000000000","tz":-300},"subject":"Switch to 2023.2 testing runtime py version","message":"Switch to 2023.2 testing runtime py version\n\nWe have prepared general testing template now and\nthat has been updated for 2023.2 (current dev cycle)\ntesting runtime.\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/873051\n\nSwitching election repo testing to the general template\nand we do not need to update it in future cycle (all adjustment\nof future cycle testing runtime will be handled by central place).\n\nChange-Id: I1ea1e98ba4b7214b4576ecac1e2fdd66bd218c06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7854ed71c7b13a7ef6c6d6e7b13522e9bd1ab63a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7854ed71c7b13a7ef6c6d6e7b13522e9bd1ab63a"}]},"branch":"refs/heads/master"},"7f56f206e623fc6d984f4737d5c85bb38beb80cc":{"kind":"REWORK","_number":2,"created":"2023-05-10 15:29:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/38/881138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/38/881138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/38/881138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/38/881138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/38/881138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/38/881138/2"}}},"commit":{"parents":[{"commit":"349b659a038993b241b17b0514516d49f3d2c592","subject":"Bobcat Election Results (TC/PTL)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/349b659a038993b241b17b0514516d49f3d2c592"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-04-20 19:25:09.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-10 15:29:50.000000000","tz":0},"subject":"Switch to 2023.2 testing runtime py version","message":"Switch to 2023.2 testing runtime py version\n\nWe have prepared general testing template now and\nthat has been updated for 2023.2 (current dev cycle)\ntesting runtime.\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/873051\n\nSwitching election repo testing to the general template\nand we do not need to update it in future cycle (all adjustment\nof future cycle testing runtime will be handled by central place).\n\nChange-Id: I1ea1e98ba4b7214b4576ecac1e2fdd66bd218c06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7f56f206e623fc6d984f4737d5c85bb38beb80cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7f56f206e623fc6d984f4737d5c85bb38beb80cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
