)]}'
{"id":"openstack%2Felection~887668","triplet_id":"openstack%2Felection~master~Ib4a2ce1e4ee74e8e9a49975017e14172b4d1f576","project":"openstack/election","branch":"master","topic":"tooling-updates","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-07-31 18:46:32.000000000","reason":"Change was submitted"},"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2023-07-16 06:51:16.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}},"hashtags":[],"change_id":"Ib4a2ce1e4ee74e8e9a49975017e14172b4d1f576","subject":"Exclude projects under the distributed leadership model","status":"MERGED","created":"2023-07-05 11:32:42.000000000","updated":"2023-07-31 18:47:27.000000000","submitted":"2023-07-31 18:46:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887668-tooling-updates","meta_rev_id":"2ebe3974f6d0d12d79d374d47b936328d04087ea","_number":887668,"virtual_id_number":887668,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-31 18:46:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-07-16 06:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2023-07-12 15:52:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2023-07-31 18:42:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-05 11:39:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-11 06:07:19.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2023-07-12 15:52:56.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"}],"messages":[{"id":"f0a6d14a375de6747c09627a5e67b485b7523a7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-05 11:32:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ce2093f9392164fe4fc11676cc574d0520781d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 11:39:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4b300b0d38478eae912160ad8faf35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a8713d52cba4dd1827a0d0d7f3793f8 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad294d54d5f6436383122763ce9792f7 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb989604b9e4409da2305ead96d6b635 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1720d75a9eec4912aca428cf21893525 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9add631947c34a6c8e8cfafc16db9b4a : SUCCESS in 4m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd0f175b4c3745b2b928d4fcb1541aeb : SUCCESS in 3m 00s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8ab7c095ec054da3aabab9232cb76903 : SUCCESS in 3m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/055135978c874d25be395e9c4700c6da : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"00189e7b3b394dd7dd3e06713110570478c2fb29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-05 12:19:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c4f8077a8e7c159ab8e7d41b2d16751539c21ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 12:28:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/151eade3953748f6a940f749f637a9ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd0d742431849b4a7295f343c9d0715 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c066a7517de48a9b9d7c0331ef1f8f2 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9cdb9f12b814f9485c675cee39f11d2 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc40fa41663f4fdd8ce023c704e9eb62 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e2c650a549f4905ac950c23450b942c : SUCCESS in 2m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c93ada3afd8043ebb09988f3b12e6749 : SUCCESS in 2m 31s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/030ab8d96e8740e2993c0ef828faca18 : SUCCESS in 2m 58s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/89a09a3cdd8c4978a9dbbc5246fb82ff : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3680638feb39f62f23a436be922beb1d852176a9","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-07-11 06:07:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f8f195e2dea568bbf1a82d309c61e96dca7bef41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-12 06:12:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"abbb5b5e8da39e9b25db2a9dfe1132a3cabb0c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 06:24:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a99e412ceca46b6af1d5b5a10ea79a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c504e191a54edbab107cf8eb9759c5 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d06423190d44f438bf65bbb19d76b4d : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c26758ac4ae42ddbce07b33cc392967 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/039838fa9ac84281b5ec4ea7e6391828 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/703ae6741cb84711a49008321f370797 : SUCCESS in 3m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/57702d6de7ac480092966a002a58be42 : SUCCESS in 3m 43s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/998c24884db54c21bbdce76a2c224843 : SUCCESS in 2m 46s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/6a820585a002456ead0edcda2a2635d2 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":3},{"id":"d4de42cab65090a822bfdd8343866b184b21b450","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2023-07-12 15:52:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7af24cf3515600f6b21864b35dd117ed34620579","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-07-16 06:51:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9b824ab3830ab6bc279db41b9045723950aa742e","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-07-31 18:42:24.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1b8220ab6481dcd11065610c55877a5ad2c83455","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 18:42:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"40394f283d3bcd67f9d86c3ef223ae8c9ad64052","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 18:46:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3fd251d8548440784db7bf6ecde34d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6deb6271e7b4d799e8782a3d3ac2a7a : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf58870aad847aeb2df5688dab2c0ab : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d529ea5ade494e79834e907d9c0adf5c : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0693f54f31a94a33b39fe0ae477dd2a2 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98f41f55c9054b769004e656b81139ef : SUCCESS in 3m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d9d6dd306674568a6a7b915842b86cc : SUCCESS in 2m 40s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/4f91b168159e49ef8bfb2a186ffc273a : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"28870cb40950f2136ca2665fdb063d1ac8a617b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 18:46:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2ebe3974f6d0d12d79d374d47b936328d04087ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 18:47:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bdf6f63181a40ec91183ca25965ab25\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/e0ece9cf6e4d412682e143c2148ede26 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eb4636f483854bc11c864d3ca1bea87f3f5113a0","revisions":{"7776146454a294c1c718b42ed20ba529a1f650d3":{"kind":"REWORK","_number":1,"created":"2023-07-05 11:32:42.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/68/887668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/887668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/887668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/887668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/887668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/887668/1"}}},"commit":{"parents":[{"commit":"671c88aab34e06cdd09fc501f287edbd608cdb45","subject":"Adding dates for upcoming election to be displayed on the election page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/671c88aab34e06cdd09fc501f287edbd608cdb45"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-30 04:00:42.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-30 04:20:44.000000000","tz":600},"subject":"Exclude projects under the distributed leader model","message":"Exclude projects under the distributed leader model\n\nCurrently there are 2 leadership types \"distributed\" and the\nDefault/unspecified which we treat as the long established PTL model.\nProjects that have selected to use the \"distributed\" model do not\nrequire a PTL election and in fact projects under this model should\nreject a PTL nomination as the correct process involves a change the\ngovernance repo.\n\nThis change updates \u0027create-directories\u0027 to exclude those projects,\nand also updates the ci checks to fail PTL nominations for \"distributed\"\nprojects.\n\nChange-Id: Ib4a2ce1e4ee74e8e9a49975017e14172b4d1f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7776146454a294c1c718b42ed20ba529a1f650d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7776146454a294c1c718b42ed20ba529a1f650d3"}]},"branch":"refs/heads/master"},"4fe8fac86975ddd1a3cd257836ce8eeff74d6a63":{"kind":"REWORK","_number":2,"created":"2023-07-05 12:19:24.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/68/887668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/887668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/887668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/887668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/887668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/887668/2"}}},"commit":{"parents":[{"commit":"671c88aab34e06cdd09fc501f287edbd608cdb45","subject":"Adding dates for upcoming election to be displayed on the election page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/671c88aab34e06cdd09fc501f287edbd608cdb45"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-30 04:00:42.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-05 12:18:57.000000000","tz":600},"subject":"Exclude projects under the distributed leader model","message":"Exclude projects under the distributed leader model\n\nCurrently there are 2 leadership types \"distributed\" and the\nDefault/unspecified which we treat as the long established PTL model.\nProjects that have selected to use the \"distributed\" model do not\nrequire a PTL election and in fact projects under this model should\nreject a PTL nomination as the correct process involves a change the\ngovernance repo.\n\nThis change updates \u0027create-directories\u0027 to exclude those projects,\nand also updates the ci checks to fail PTL nominations for \"distributed\"\nprojects.\n\nChange-Id: Ib4a2ce1e4ee74e8e9a49975017e14172b4d1f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4fe8fac86975ddd1a3cd257836ce8eeff74d6a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4fe8fac86975ddd1a3cd257836ce8eeff74d6a63"}]},"branch":"refs/heads/master"},"eb4636f483854bc11c864d3ca1bea87f3f5113a0":{"kind":"REWORK","_number":3,"created":"2023-07-12 06:12:27.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/68/887668/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/887668/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/887668/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/887668/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/887668/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/887668/3"}}},"commit":{"parents":[{"commit":"671c88aab34e06cdd09fc501f287edbd608cdb45","subject":"Adding dates for upcoming election to be displayed on the election page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/671c88aab34e06cdd09fc501f287edbd608cdb45"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-30 04:00:42.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-12 05:15:24.000000000","tz":600},"subject":"Exclude projects under the distributed leadership model","message":"Exclude projects under the distributed leadership model\n\nCurrently there are 2 leadership types \"distributed\" and the\nDefault/unspecified which we treat as the long established PTL model.\nProjects that have selected to use the \"distributed\" model do not\nrequire a PTL election and in fact projects under this model should\nreject a PTL nomination as the correct process involves a change the\ngovernance repo.\n\nThis change updates \u0027create-directories\u0027 to exclude those projects,\nand also updates the ci checks to fail PTL nominations for \"distributed\"\nprojects.\n\nChange-Id: Ib4a2ce1e4ee74e8e9a49975017e14172b4d1f576\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/eb4636f483854bc11c864d3ca1bea87f3f5113a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/eb4636f483854bc11c864d3ca1bea87f3f5113a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
