)]}'
{"id":"openstack%2Felection~887669","triplet_id":"openstack%2Felection~master~Ie7a9ffcdbe6a28cd2a80289ec06259ec869d306f","project":"openstack/election","branch":"master","topic":"tooling-updates","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-07-31 18:51:16.000000000","reason":"Change was submitted"},"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2023-07-16 06:56:26.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},"15993":{"account":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"last_update":"2023-07-31 18:51:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7a9ffcdbe6a28cd2a80289ec06259ec869d306f","subject":"Add a Sorting function.","status":"MERGED","created":"2023-07-05 11:32:42.000000000","updated":"2023-07-31 18:52:08.000000000","submitted":"2023-07-31 18:51:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887669-tooling-updates","meta_rev_id":"278a9c7d040eb55f2ba3c3b5bdfc6e1eb0acaa70","_number":887669,"virtual_id_number":887669,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-31 18:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-07-16 06:56:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2023-07-31 18:42:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-05 11:49:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-11 05:10:48.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2023-07-11 23:41:39.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2023-07-12 15:53:29.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"}],"messages":[{"id":"1d20238d2de9ff168a766a8540369eb09babf1ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-05 11:32:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f27f2d3f510017423277400635a401ccae30550","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 11:49:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/920ab1ef34314031be90c4647cc8b13b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a95b374b7f470880003c44bf40e16e : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/364f1cc305d34ef69a4d98280f99a180 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8e7a3ab326142039be0c75d83e6bde5 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6a9a09426504ab090bee7ec398bd991 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/328cd56292804fa4a7ccbbf758bcd6b6 : SUCCESS in 3m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a1212b273794cb48a04b99df3b31a5a : SUCCESS in 2m 50s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/0c565f291f6e43e4a4ae42aaa94469f1 : SUCCESS in 3m 06s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5a383ab8905f4dfe9259f4e297213f89 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"24b29bde3ad3708522288b4f8bf387d09076819a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-05 12:19:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dc522782ed90fc5a9eb9dd4b699b63c30efd6447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 12:25:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be03705f201c41bcafc6cf929ad4872b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bea8ef2219d4a36bb32722b48258869 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d76d94a563a45b1af8eb1774fdbd28a : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbedafb8eeba4cd1b5a8dd8afedc3773 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b8941d0ea524e33b2d55371a7be8a1a : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4470682f6d6045aeb9fa98664d44d173 : SUCCESS in 3m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6cb1e8b151944a47838c8f3973127e70 : SUCCESS in 3m 26s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/328db9e02e5e4893b74a6b05fd7da200 : SUCCESS in 2m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/34ee154e9a3b4e7d8ab65561a6005664 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"12a302f6b3cfb668836c5120a80750293d9acb48","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-07-11 05:10:48.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0720708b9249ae9027eb56f57231485b5c2bfeb9","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-11 05:16:00.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cd27d4ec6caba69aeec7212a44c771031bc19b1a","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-07-11 23:41:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ee674e97ec669c7fbfe975109a50d4b93bf1cfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-12 06:12:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cba926cdc60f154d694c46114e95941434c57c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 06:20:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebbcbc8d63f44d3da9155b21b74ddd21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9101598900364faabd1f452fe7c0c78b : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8059f42b6c594744a0e9fc3c29c00bd7 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22eb057b31794435bcafdae443dae172 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aefb01029d684243ad8610eb6fe27bdf : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07fb789a27254ab892257d30cd4ea81c : SUCCESS in 4m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6736414130ca4a37813bdfcbefac4348 : SUCCESS in 3m 05s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/0b3c3104225c405ebfab2a1ffa60c08a : SUCCESS in 2m 35s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5ba0327fdcee45b98b8e6a9b3f7621c6 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"264f8bb2f49f3a4bbdc0f73fc37d852e102c106b","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2023-07-12 15:53:29.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"25eb238e9ed0b1bbb5b35124dc736d8c5846e040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-13 06:41:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ca871f889427486c28cfd44f8462a0f8ef2d7f2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 06:50:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eff246459be42a7bbd8d8fd2e58e54b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90326d8d0b1d497dba1dfd5785491f1a : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4604e1798e5449c297f8c311d29b40a1 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/886c1b41b854425287b654099d247a9c : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a3cdead860d448b92991a931c2c8d25 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d80e81a6f4e84ce2b4eb04241f9fd1d4 : SUCCESS in 2m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/08c64e4bce2043098dd7d7bad4ec43d8 : SUCCESS in 2m 48s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7a067bfe463e4d9f8a9dbc2a9af35f28 : SUCCESS in 2m 51s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9384d83302c84fd3a4fc8b728a69be4d : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":4},{"id":"62efdbb84788f9b1897ec580be3d8812f41aedf7","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-07-16 06:56:26.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f0d52dc1b29bb7c893e79365e531cdbd009d5cf","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-07-31 18:42:14.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2b800a68c2013cd8b92bbb30802e3cb136b9c9d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 18:42:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7d13fd4c1cb9d9661a1f6eb6f328b15d91616182","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 18:51:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9115c2d5874e4bcdb30e058bbabaa800\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69fadc7de2e54457857143fc0deb40c0 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c31df38f5144a5973052543aced65a : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f1449e974ab4328870262afa0beff0e : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b4e96864588496b8f96f4a16177ef78 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/934f01aa98484296831d9c69aac43fd6 : SUCCESS in 3m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4e92a8208c1b4023ad9a0d0210948b7a : SUCCESS in 2m 34s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/79fd263d24b54a269ea51f580f93770d : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":4},{"id":"cc21eb9900be41b9298af7eed1e5a598244f300b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 18:51:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"278a9c7d040eb55f2ba3c3b5bdfc6e1eb0acaa70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 18:52:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce3c0e9c635b462f84ddbdb1c329596c\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/cd9359d5438d4e5981167c73fd159b70 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cb661a42f823741c1e6b10f7d3b6afda2f81029f","revisions":{"fd6ced3ad9433f06c1655015e0b3c06e8c904b6f":{"kind":"REWORK","_number":1,"created":"2023-07-05 11:32:42.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/69/887669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/69/887669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/69/887669/1"}}},"commit":{"parents":[{"commit":"7776146454a294c1c718b42ed20ba529a1f650d3","subject":"Exclude projects under the distributed leader model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7776146454a294c1c718b42ed20ba529a1f650d3"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-05 08:48:52.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-05 11:17:10.000000000","tz":600},"subject":"Add a Sorting function.","message":"Add a Sorting function.\n\nCurrently we\u0027re reverse sorting with a simple sort, this sadly puts\nnewer released (antalope, bobcat) at the end of the list.  Add a\nkey function that has context around our release names and release ids\nto sort \"2024.1/cantaloupe\" at the top.\n\nChange-Id: Ie7a9ffcdbe6a28cd2a80289ec06259ec869d306f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fd6ced3ad9433f06c1655015e0b3c06e8c904b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fd6ced3ad9433f06c1655015e0b3c06e8c904b6f"}]},"branch":"refs/heads/master"},"e73682ca158729a35266dee53285089e3d4a9c8a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-05 12:19:24.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/69/887669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/69/887669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/69/887669/2"}}},"commit":{"parents":[{"commit":"4fe8fac86975ddd1a3cd257836ce8eeff74d6a63","subject":"Exclude projects under the distributed leader model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4fe8fac86975ddd1a3cd257836ce8eeff74d6a63"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-05 08:48:52.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-05 12:18:57.000000000","tz":600},"subject":"Add a Sorting function.","message":"Add a Sorting function.\n\nCurrently we\u0027re reverse sorting with a simple sort, this sadly puts\nnewer released (antalope, bobcat) at the end of the list.  Add a\nkey function that has context around our release names and release ids\nto sort \"2024.1/cantaloupe\" at the top.\n\nChange-Id: Ie7a9ffcdbe6a28cd2a80289ec06259ec869d306f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e73682ca158729a35266dee53285089e3d4a9c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e73682ca158729a35266dee53285089e3d4a9c8a"}]},"branch":"refs/heads/master"},"50de3d12fcbe459f14e84a39b8dd33c980a83f3c":{"kind":"REWORK","_number":3,"created":"2023-07-12 06:12:27.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/69/887669/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/69/887669/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/69/887669/3"}}},"commit":{"parents":[{"commit":"eb4636f483854bc11c864d3ca1bea87f3f5113a0","subject":"Exclude projects under the distributed leadership model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/eb4636f483854bc11c864d3ca1bea87f3f5113a0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-05 08:48:52.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-12 06:06:19.000000000","tz":600},"subject":"Add a Sorting function.","message":"Add a Sorting function.\n\nCurrently we\u0027re reverse sorting with a simple sort, this sadly puts\nnewer released (antalope, bobcat) at the end of the list.  Add a\nkey function that has context around our release names and release ids\nto sort \"2024.1/cantaloupe\"[1] at the top.\n\n[1] As of now the c name isn\u0027t selected cantaloupe is my placeholder\n\nChange-Id: Ie7a9ffcdbe6a28cd2a80289ec06259ec869d306f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/50de3d12fcbe459f14e84a39b8dd33c980a83f3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/50de3d12fcbe459f14e84a39b8dd33c980a83f3c"}]},"branch":"refs/heads/master"},"cb661a42f823741c1e6b10f7d3b6afda2f81029f":{"kind":"REWORK","_number":4,"created":"2023-07-13 06:41:13.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/69/887669/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/69/887669/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/69/887669/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/69/887669/4"}}},"commit":{"parents":[{"commit":"eb4636f483854bc11c864d3ca1bea87f3f5113a0","subject":"Exclude projects under the distributed leadership model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/eb4636f483854bc11c864d3ca1bea87f3f5113a0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-05 08:48:52.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-13 06:39:42.000000000","tz":600},"subject":"Add a Sorting function.","message":"Add a Sorting function.\n\nCurrently we\u0027re reverse sorting with a simple sort, this sadly puts\nnewer released (antalope, bobcat) at the end of the list.  Add a\nkey function that has context around our release names and release ids\nto sort \"2024.1/cantaloupe\"[1] at the top.\n\n[1] As of now the c name isn\u0027t selected cantaloupe is my placeholder\n\nChange-Id: Ie7a9ffcdbe6a28cd2a80289ec06259ec869d306f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/cb661a42f823741c1e6b10f7d3b6afda2f81029f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/cb661a42f823741c1e6b10f7d3b6afda2f81029f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
