)]}'
{"id":"openstack%2Felection~888209","triplet_id":"openstack%2Felection~master~Icc67336742f43030d4e1743c6db691fafc250cfa","project":"openstack/election","branch":"master","topic":"tooling-updates","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2024-09-19 18:40:21.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-09-19 18:40:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc67336742f43030d4e1743c6db691fafc250cfa","subject":"Simplify and update-governance","status":"MERGED","created":"2023-07-12 06:12:27.000000000","updated":"2024-09-19 18:41:24.000000000","submitted":"2024-09-19 18:40:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"888209-tooling-updates","meta_rev_id":"6905891a39c12bc3e120141bf155dfd1f8adc9b7","_number":888209,"virtual_id_number":888209,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-19 18:40:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-09-19 14:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2024-09-19 18:31:43.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2024-09-19 18:31:43.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-12 06:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-19 14:24:31.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2024-09-19 18:31:43.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"6ca302292a2d6c18cde0d122d3a7f93a285525c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-12 06:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"128e3357b10ece0031f83c8e08b3424102e0035c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 06:21:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/012bdef4d5244726b1669a5b3e6bd2cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7289a4e08932465b949d05494fc97295 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f776d171c0454fab9af890afa99b1f93 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeffbc704ba44e47a1e35a82d18bfa1f : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/072af635b72e443487806481b3a84df4 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33a7c3c774564f6b97ff457c8beb98d7 : SUCCESS in 2m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/41cfc3081f084af3b964f97114474297 : SUCCESS in 3m 55s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/81b6293d6ead40c0b55bd0a99c5fc97a : SUCCESS in 2m 40s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/716f776f94cf49f4904b07c4238283ab : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1c523c8cb4dc1a85506cf6d67541118c2cbdfeed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-13 06:41:13.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0bef3cd6116331a30a9f789db0b8f0f4f10f50f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 06:49:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71b5948dd5724cf5bec497205454eee1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9def7cfe7bb643f786ae0e95d9815fb0 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18af34c9f1dd4ee1b2efbb467c7dc081 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82fb9eb8f3a34fecb26a2bec21421d2c : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60f949999cf74ff6bfe4e1d405c97c2a : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cca09fa0fac4187aaad29f7e949a892 : SUCCESS in 3m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cc73e61c62514d9a8dae203f681564be : SUCCESS in 3m 00s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/cb672eb8b2bc4ff68dbba9c1a1b23ee4 : SUCCESS in 3m 02s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/bc4644d16f794874b93e3d55ad6995c7 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"081fed9216c53a961258297d8f18a69a4c359b05","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-07-31 19:01:08.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61a1733a5a5b78fa9034f07710538c61d1bd48ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-09-21 09:34:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"530fd85c7bacce87f97793a545a5187e5f448920","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-09-21 09:35:37.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f552dfd7d04cb176c21aaf6f9b390b38f8380ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 09:44:44.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f1de8c13e9742cf8d95620c618051e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed94ef93c2e4bae9b0cb37e1d775968 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b9b930ff3174aa1afea987af741259e : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac1d88ecc3b14f2d9c5ed699f4079a3a : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28b7ed3816ad4393aad484aaf99b242a : SUCCESS in 2m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4665449080ed471d828dbda40a9f0a54 : FAILURE in 4m 02s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/f09e34fc163845f496b9dafc37d4aacd : SUCCESS in 3m 01s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/dc9d61c6549a4a1db4e61609d68cba86 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":3},{"id":"7cd814ef1a69427961059966478e96929f4356e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-03-06 04:05:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c7f3b39fc93017ab0922f36045812140614cf4f6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-03-06 04:09:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6da72b13ef3a446cae1ec7ca7f848631e0b10ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 04:14:21.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edbb471ceb8c47308bfd2056e4e5fd82\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d493ef2b30e41fd903c131db2d3836a : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/180ebecbb06b48e4a02b63109a63d284 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8035d17916b9462e8921ba052c54b1f4 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3859afd6972b4484add9080a62f9cbc0 : SUCCESS in 3m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/499a70c62d824ad0838926cdb2e60bc5 : FAILURE in 2m 38s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/f003e64c524f441aaa2b30eeba7d15dd : SUCCESS in 2m 56s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/abe0b21170f243018ee77ad2490fc590 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":4},{"id":"d01c74e37ac2b08c2e433fd0af78e31ff40e83fb","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-08-08 00:08:00.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"6cba0cf3f06712ddd98efe03219e06ad4f94a50e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-08-08 00:08:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aecdcd833113bac46695a3494b97ee1db1d37ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-09-19 14:19:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"31d1bedfcd54edefab6e409514751e733133b998","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-09-19 14:24:31.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"53dac5a4d7eb5741e54f130bc6eb28e3f603b569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 14:28:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0916089e67914699bf25597f8b7e3acf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ccd3e42e7843868b7b21d9100b7f0a : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3aa4b404af54c8c818f2c2f9cd398dd : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/442789ac26bc49efb0e0315db5cac276 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efab99fd8da84bf7abc9016880ac5f68 : SUCCESS in 2m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/42a5aded9a6b4e51ac2abdacbdee3c6b : SUCCESS in 2m 53s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7b46dd55b4de4bdcb87140335efa4de9 : SUCCESS in 2m 26s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f847347498324e1e8593daf3f7015f4c : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":5},{"id":"4410b6fb25396f7875344e44c57694104ad652a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-09-19 14:40:36.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fe49c363a011ea92658047dbd28ade543dd9a25a","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-09-19 14:43:15.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ed1f79c1082e99970d3695971151fcbb3ac2efea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 14:53:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e412f16a84b449aa2143ce81a6c544f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3929453fc312443aae6977377128053a : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41b65ea34aab4fa2b030b6fd26403d86 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6204944dbda14c6d85c75f5eadd895d5 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6754978a29c34929a4afeadb11edad08 : SUCCESS in 3m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9f0e7206c665469794acc66644434601 : SUCCESS in 3m 14s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/19dcda61066a45af85d2a7fdaca5ab60 : SUCCESS in 2m 52s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1433ceb37c904402b447907fcb7fcc8f : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":6},{"id":"b77a0aed3bdfa6af4311767ccbacf6b537d15625","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-19 18:31:43.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4a2e771a6c045554b19567adaf133e8cf1d6b43b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 18:32:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1362293f2bae11765a3f5b1ed2c2e8afac9d23d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 18:40:21.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff8d22331c1b4a4e847d024784440d2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1d14171ae348d39eadb64551e17a70 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/517f39aee8a2432b8d95c3c5c8b673d9 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae96fb48700c45cca3874596a913953f : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15c11a19a3d5434b97866b6bd9b9352e : SUCCESS in 2m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e45f36fc1d714ca5b2694fac39955f91 : SUCCESS in 2m 39s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8a45f84d265b4e89aa2ee13e34c63dc8 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":6},{"id":"c91a0cee8b7ef6b7883e610a183f94ab571bf01d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 18:40:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6905891a39c12bc3e120141bf155dfd1f8adc9b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 18:41:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00a0795bd7674cfb8e93b9a2d7e3f515\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/d49449e74709493a93fe20687e1f6bf1 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"dae72c05279c0e4f818c56e51474d2335265e46a","revisions":{"c0112b253d772120fce97204619049ed3ff7c49b":{"kind":"REWORK","_number":1,"created":"2023-07-12 06:12:27.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/09/888209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/09/888209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/09/888209/1"}}},"commit":{"parents":[{"commit":"bd08b9e3201059b6ef2ab57cd0eec14f593d12f6","subject":"Add a tool to update the releases repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/bd08b9e3201059b6ef2ab57cd0eec14f593d12f6"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-12 05:09:57.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-12 06:07:49.000000000","tz":600},"subject":"Simplify and update-governance","message":"Simplify and update-governance\n\nThe existing version does not handle projects under the \"distributed\" leadership model, if also badly damages projects.yaml and does unneeded, potentially error prone, work to determine the election results\n\nSimplify the tool to address these issues\n\nChange-Id: Icc67336742f43030d4e1743c6db691fafc250cfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c0112b253d772120fce97204619049ed3ff7c49b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c0112b253d772120fce97204619049ed3ff7c49b"}]},"branch":"refs/heads/master"},"c9ef0be8153c4cbe056ca6b900b05ae780ee2397":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-13 06:41:13.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/09/888209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/09/888209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/09/888209/2"}}},"commit":{"parents":[{"commit":"056184c363e056e82189551454e543b81afb9a33","subject":"Add a tool to update the releases repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/056184c363e056e82189551454e543b81afb9a33"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-12 05:09:57.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-13 06:40:20.000000000","tz":600},"subject":"Simplify and update-governance","message":"Simplify and update-governance\n\nThe existing version does not handle projects under the \"distributed\" leadership model, if also badly damages projects.yaml and does unneeded, potentially error prone, work to determine the election results\n\nSimplify the tool to address these issues\n\nChange-Id: Icc67336742f43030d4e1743c6db691fafc250cfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c9ef0be8153c4cbe056ca6b900b05ae780ee2397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c9ef0be8153c4cbe056ca6b900b05ae780ee2397"}]},"branch":"refs/heads/master"},"bdb95f2cd294e6a985b8fc56527cf0fa4e04588f":{"kind":"REWORK","_number":3,"created":"2023-09-21 09:34:03.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/09/888209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/09/888209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/09/888209/3"}}},"commit":{"parents":[{"commit":"99421c3568e245da8b364f79022114caaae24ed7","subject":"Add 2024.1/Caracal results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/99421c3568e245da8b364f79022114caaae24ed7"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-12 05:09:57.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-09-21 09:33:41.000000000","tz":600},"subject":"Simplify and update-governance","message":"Simplify and update-governance\n\nThe existing version does not handle projects under the \"distributed\" leadership model, if also badly damages projects.yaml and does unneeded, potentially error prone, work to determine the election results\n\nSimplify the tool to address these issues\n\nChange-Id: Icc67336742f43030d4e1743c6db691fafc250cfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/bdb95f2cd294e6a985b8fc56527cf0fa4e04588f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/bdb95f2cd294e6a985b8fc56527cf0fa4e04588f"}]},"branch":"refs/heads/master"},"73ed3f46538d74a8e397c05cf3114aa1c9156b5c":{"kind":"REWORK","_number":4,"created":"2024-03-06 04:05:19.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/09/888209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/09/888209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/09/888209/4"}}},"commit":{"parents":[{"commit":"64e7417bccdbb85fecd9147fe8a87ee8941ac080","subject":"[templates] Update links","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/64e7417bccdbb85fecd9147fe8a87ee8941ac080"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-12 05:09:57.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-03-06 04:05:07.000000000","tz":660},"subject":"Simplify and update-governance","message":"Simplify and update-governance\n\nThe existing version does not handle projects under the \"distributed\" leadership model, if also badly damages projects.yaml and does unneeded, potentially error prone, work to determine the election results\n\nSimplify the tool to address these issues\n\nChange-Id: Icc67336742f43030d4e1743c6db691fafc250cfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/73ed3f46538d74a8e397c05cf3114aa1c9156b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/73ed3f46538d74a8e397c05cf3114aa1c9156b5c"}]},"branch":"refs/heads/master"},"b7b9997f8309a20c5cfcc43d301da8922a32225f":{"kind":"REWORK","_number":5,"created":"2024-09-19 14:19:18.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/09/888209/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/09/888209/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/09/888209/5"}}},"commit":{"parents":[{"commit":"1e77196ddf40ac688fd292da84e3573977feef5b","subject":"Withdrawing Walt Boring Candidacy for cinder PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1e77196ddf40ac688fd292da84e3573977feef5b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-12 05:09:57.000000000","tz":600},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-09-19 14:14:52.000000000","tz":540},"subject":"Simplify and update-governance","message":"Simplify and update-governance\n\nThe existing version does not handle projects under the \"distributed\"\nleadership model, if also badly damages projects.yaml and does unneeded,\npotentially error prone, work to determine the election results.\n\nSimplify the tool to address these issues\n\nChange-Id: Icc67336742f43030d4e1743c6db691fafc250cfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b7b9997f8309a20c5cfcc43d301da8922a32225f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b7b9997f8309a20c5cfcc43d301da8922a32225f"}]},"branch":"refs/heads/master"},"dae72c05279c0e4f818c56e51474d2335265e46a":{"kind":"REWORK","_number":6,"created":"2024-09-19 14:40:36.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/09/888209/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/09/888209/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/09/888209/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/09/888209/6"}}},"commit":{"parents":[{"commit":"1e77196ddf40ac688fd292da84e3573977feef5b","subject":"Withdrawing Walt Boring Candidacy for cinder PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1e77196ddf40ac688fd292da84e3573977feef5b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-07-12 05:09:57.000000000","tz":600},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-09-19 14:40:25.000000000","tz":540},"subject":"Simplify and update-governance","message":"Simplify and update-governance\n\nThe existing version does not handle projects under the \"distributed\"\nleadership model, if also badly damages projects.yaml and does unneeded,\npotentially error prone, work to determine the election results.\n\nSimplify the tool to address these issues\n\nChange-Id: Icc67336742f43030d4e1743c6db691fafc250cfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dae72c05279c0e4f818c56e51474d2335265e46a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dae72c05279c0e4f818c56e51474d2335265e46a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
