)]}'
{"id":"openstack%2Felection~891131","triplet_id":"openstack%2Felection~master~Ib9732fa5f834ddf84d6f8526c1af4128f011f7d2","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-08-11 22:36:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9732fa5f834ddf84d6f8526c1af4128f011f7d2","subject":"Ignore tox4 .eggs directory","status":"MERGED","created":"2023-08-10 20:58:57.000000000","updated":"2023-08-11 22:36:51.000000000","submitted":"2023-08-11 22:36:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891131","meta_rev_id":"2bebe674221352f45f9dd5a0f8930d619dc7b649","_number":891131,"virtual_id_number":891131,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-11 22:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-08-11 21:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-11 18:31:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2023-08-11 21:01:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-10 21:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-11 18:24:11.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-08-11 21:01:05.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"be01f9cd687d728d7778926ffc4a64ff85097d83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-10 20:58:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79d9b2e7026f850288c267a64e3065cea119e10e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-10 21:05:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c0319e1a05f6c3df943b548064771cf09fb119cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 21:15:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88bd1e9206214d819d378421c568bd51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93c423dedd94b67bfbedcf3db2ae578 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b0309d9dfaa4608be7ad54221236d2a : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/675d4fdf84014642b6eb7a13d32d132a : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e15fa2e48324449bb60d6986dbab4bd : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1225f3d5f4c340caad4197b775190082 : SUCCESS in 3m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/094653e435f341fcb6c464156e11dfb5 : SUCCESS in 3m 07s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e5c7dcc8abf5434884338c47cfc4c3c2 : SUCCESS in 3m 34s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/896eb3f62d044d4da4500532ed3f1a10 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"26e7e9db3b8fb1b651db21a60c50eb1225b66f32","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-08-11 18:24:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c3833aa395bcef0fdfdca7a5eab344e5565e843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-11 18:27:15.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"70f8354e04ff5fee720c193d7e2e56709f6947ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-11 18:31:09.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ae60aedaf8d7ab5199d6894ac137c5bdd2607aea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 18:40:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f8290f90995431fb052a47240b2b29a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b9cc522fa4d479189e6df8abdc2f799 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45acfc9fe69a45d9b72dd99c3022c891 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d23a41b3cfa46baa5e74968a1c08f35 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/336097e59604408ea586616106bc0ada : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f77c830ffe342d9a0c48acaea56872c : SUCCESS in 6m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4ce87d2800ca440c88cb1bb6b836a4cd : SUCCESS in 3m 01s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/811b3c06b16f48879ecf5ec9297fa791 : SUCCESS in 3m 09s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/992dfa60f8ab402d9c35a190ce0cd514 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":4},{"id":"3fa475a62db4c27d26f5ae32391ff549a847b673","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-11 21:01:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4cea14f5f77fb6e7002470852c5893d4afe64cf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 22:22:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fa34b15854eba7e24076d62eb50898db2c3d10ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 22:36:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/686ae2a930f047898ed096cac294533f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6568855a56084450b97375ce65e21eee : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/632c6496e17e48bb9ea794a3d3c6b4e9 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c900d033833429ebb00aa486e5ec493 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5c3d60108404f399003031cc70d57f1 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f1bb325afdc4456a92decd9677c0ced : SUCCESS in 2m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa5749bcc4614ae59a4c64cfa003a3a3 : SUCCESS in 3m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c7525ebe02c84256bd51293e99bb7cbc : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":4},{"id":"2aaaab3a565c48fcdab2d671fec70f19edad2ec1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 22:36:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2bebe674221352f45f9dd5a0f8930d619dc7b649","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 22:36:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85a4b28228647659d4f60b7cfac83fe\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/450eb13351174a5e883d150d443b3ca0 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"afbb77a435d443206505e4577d4452799e163832","revisions":{"0d5af5359c83dfa3c59a1512774ef19c51539f39":{"kind":"REWORK","_number":1,"created":"2023-08-10 20:58:57.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/31/891131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/31/891131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/31/891131/1"}}},"commit":{"parents":[{"commit":"dd489c23055081b4395f80e8608461582ee1e3fa","subject":"Add process documentation for update-release-calendar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dd489c23055081b4395f80e8608461582ee1e3fa"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 20:58:20.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 20:58:20.000000000","tz":-300},"subject":"Ignore tox4 .eggs directory","message":"Ignore tox4 .eggs directory\n\nChange-Id: Ib9732fa5f834ddf84d6f8526c1af4128f011f7d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0d5af5359c83dfa3c59a1512774ef19c51539f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0d5af5359c83dfa3c59a1512774ef19c51539f39"}]},"branch":"refs/heads/master"},"1a1e4c474f9b8585f36c13b651bdf6d1e0e85ae8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-10 21:05:43.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/31/891131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/31/891131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/31/891131/2"}}},"commit":{"parents":[{"commit":"a6f0cfcb07733197a4007ae8f22ea0572691f0f0","subject":"Add process documentation for update-release-calendar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a6f0cfcb07733197a4007ae8f22ea0572691f0f0"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 20:58:20.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 21:05:32.000000000","tz":-300},"subject":"Ignore tox4 .eggs directory","message":"Ignore tox4 .eggs directory\n\nChange-Id: Ib9732fa5f834ddf84d6f8526c1af4128f011f7d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1a1e4c474f9b8585f36c13b651bdf6d1e0e85ae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1a1e4c474f9b8585f36c13b651bdf6d1e0e85ae8"}]},"branch":"refs/heads/master"},"93cbc73533230970df68959dd3c57db45f9daa5e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-11 18:27:15.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/31/891131/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/31/891131/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/31/891131/3"}}},"commit":{"parents":[{"commit":"b16a26ffeb531c679533492c1d23b0ca02c52d98","subject":"Add process documentation for update-release-calendar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b16a26ffeb531c679533492c1d23b0ca02c52d98"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 20:58:20.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-11 18:25:45.000000000","tz":-300},"subject":"Ignore tox4 .eggs directory","message":"Ignore tox4 .eggs directory\n\nChange-Id: Ib9732fa5f834ddf84d6f8526c1af4128f011f7d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/93cbc73533230970df68959dd3c57db45f9daa5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/93cbc73533230970df68959dd3c57db45f9daa5e"}]},"branch":"refs/heads/master"},"afbb77a435d443206505e4577d4452799e163832":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-08-11 18:31:09.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/31/891131/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/31/891131/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/31/891131/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/31/891131/4"}}},"commit":{"parents":[{"commit":"94cf32bc49b041a0e3a197c5aefd2e6a63d2746e","subject":"Add process documentation for update-release-calendar","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/94cf32bc49b041a0e3a197c5aefd2e6a63d2746e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 20:58:20.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-11 18:31:03.000000000","tz":-300},"subject":"Ignore tox4 .eggs directory","message":"Ignore tox4 .eggs directory\n\nChange-Id: Ib9732fa5f834ddf84d6f8526c1af4128f011f7d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/afbb77a435d443206505e4577d4452799e163832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/afbb77a435d443206505e4577d4452799e163832"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
