)]}'
{"id":"openstack%2Felection~891135","triplet_id":"openstack%2Felection~master~I6eaa238ae8f223c6c76fcc53b59b335a0ba6b381","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-08-11 22:30:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6eaa238ae8f223c6c76fcc53b59b335a0ba6b381","subject":"Add combined_campaigning_kickoff.j2","status":"MERGED","created":"2023-08-10 21:05:43.000000000","updated":"2023-08-11 22:31:18.000000000","submitted":"2023-08-11 22:30:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891135","meta_rev_id":"86ba0f159a2d038c76f0a2574495ce02abca9048","_number":891135,"virtual_id_number":891135,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-11 22:30:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-08-11 22:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-11 18:31:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2023-08-11 22:22:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-10 21:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-11 18:23:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-08-11 22:22:29.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"738137ea6aca0d751268859d5242a73f2ce79994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-10 21:05:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"901afd444387c31fdde8970be789a0be0534d402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 21:16:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e41cb117b8046baaa604cd8b916cba2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9848295f7ddf45c3aa03a144adff69d0 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22091466170c4444bc346c99f4763b3f : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/974d151adfe842f89299f8f81e2c1740 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aeef6691b6f74cada19c196f83fca50c : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ec1fb9da6d84749a06b69951acee0ee : SUCCESS in 3m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fe6a968f478241d2890e3bc41a862a61 : SUCCESS in 3m 01s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/1f48267050a54a72b0c3562c7559b831 : SUCCESS in 2m 53s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8d9c6e1dcf2a4ad0b15779a24c57e73f : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"92bf6625a9e3e6a8758d65e2289e5c0aaf440524","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-08-11 18:23:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e295e1d7465d0e56282f5fbe56cbb177640b0b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-11 18:27:15.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c16957b5efa5596c26e04730e7b3993e31cab47f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-11 18:31:09.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0595789f4e11d02ea5c734e6bada2d74119356de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 18:46:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0fda3a76fa49d1817a152119872132\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2cb1d363dd47c3afb914af5dc574b0 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f7f3653c1914557bd64b224ea4f18a9 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22e02ae4deac4f03adf1f9e8652abe39 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2109b7f98ecf4d6e9e8065e36f5bc566 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1971bf56d2a941e6b040279f039e7960 : SUCCESS in 3m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2c064bfcf4ce43c7a10ad00cbc632d7a : SUCCESS in 3m 06s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/4084d61b44094f99853d0f1687796451 : SUCCESS in 2m 45s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e56fe747016b4941ac826203c7cf2450 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":3},{"id":"1a061a5aa0f7545c32110844917b4aaa89c444c9","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-11 22:22:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6031ce3dadc0b26a430891c31feac01c1f28745c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 22:22:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7d6c5997150e8892071d787f3e12eddca1b3b898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 22:30:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9602e73b941f4c769d2ba7c07f3617ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2ccf6ae16ff466f83435c205823bbc8 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bcc9c92b08748aa996c7d9caeca21e1 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5fafb2083df424abf98608c07486fa8 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d953a5df7c447ebab5ef6900345da0a : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5f4a26bbb194366b89c7eeb880df5c8 : SUCCESS in 2m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/273341cb533d40c29e5e64b2e23d3f56 : SUCCESS in 3m 29s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2b51ca2b0f25424699c6301e4bbd6b0b : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":3},{"id":"50371c89e3f7ac370b01b35066bebdd82abddb64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 22:30:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"86ba0f159a2d038c76f0a2574495ce02abca9048","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 22:31:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c93999622344c34a3a9afed23d291a4\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/7f0fc2686abc4c8f923d3b1d52cb2cb7 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b94b9820904e3794bad30c007c00cf331f2678bb","revisions":{"33e4922131da64eb3ed18c1900a5d26860752314":{"kind":"REWORK","_number":1,"created":"2023-08-10 21:05:43.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/891135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/35/891135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/35/891135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/35/891135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/35/891135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/35/891135/1"}}},"commit":{"parents":[{"commit":"12399f95fd58d10fdbb961f2a80a5e5b7bb7b92b","subject":"Update election_season season templates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/12399f95fd58d10fdbb961f2a80a5e5b7bb7b92b"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 21:04:22.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 21:05:32.000000000","tz":-300},"subject":"Add combined_campaigning_kickoff.j2","message":"Add combined_campaigning_kickoff.j2\n\nAlso while there change footnoting style\n\nChange-Id: I6eaa238ae8f223c6c76fcc53b59b335a0ba6b381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/33e4922131da64eb3ed18c1900a5d26860752314"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/33e4922131da64eb3ed18c1900a5d26860752314"}]},"branch":"refs/heads/master"},"51af57f00946b2fe4d61872f26d3bae59dffac69":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-11 18:27:15.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/891135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/35/891135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/35/891135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/35/891135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/35/891135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/35/891135/2"}}},"commit":{"parents":[{"commit":"f3bdda2855e0d2614154e68569e4632be2edf958","subject":"Update election_season season templates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/f3bdda2855e0d2614154e68569e4632be2edf958"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 21:04:22.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-11 18:25:45.000000000","tz":-300},"subject":"Add combined_campaigning_kickoff.j2","message":"Add combined_campaigning_kickoff.j2\n\nAlso while there change footnoting style\n\nChange-Id: I6eaa238ae8f223c6c76fcc53b59b335a0ba6b381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/51af57f00946b2fe4d61872f26d3bae59dffac69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/51af57f00946b2fe4d61872f26d3bae59dffac69"}]},"branch":"refs/heads/master"},"b94b9820904e3794bad30c007c00cf331f2678bb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-11 18:31:09.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/891135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/35/891135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/35/891135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/35/891135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/35/891135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/35/891135/3"}}},"commit":{"parents":[{"commit":"52680e976eaae48430fa3457388e576554050463","subject":"Update election_season season templates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/52680e976eaae48430fa3457388e576554050463"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-10 21:04:22.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-08-11 18:31:03.000000000","tz":-300},"subject":"Add combined_campaigning_kickoff.j2","message":"Add combined_campaigning_kickoff.j2\n\nAlso while there change footnoting style\n\nChange-Id: I6eaa238ae8f223c6c76fcc53b59b335a0ba6b381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b94b9820904e3794bad30c007c00cf331f2678bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b94b9820904e3794bad30c007c00cf331f2678bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
