)]}'
{"id":"openstack%2Felection~891568","triplet_id":"openstack%2Felection~master~I9d550b25ee46d9024f07095cc81d5eae8dcb3842","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-18 19:10:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9d550b25ee46d9024f07095cc81d5eae8dcb3842","subject":"[2024.1] Add mnasiadka candidacy for Kolla","status":"MERGED","created":"2023-08-16 08:19:41.000000000","updated":"2023-08-18 19:10:43.000000000","submitted":"2023-08-18 19:10:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891568","meta_rev_id":"ba3c1c3d6acecb8d88b0c6b323e5183e498e60bf","_number":891568,"virtual_id_number":891568,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-18 19:10:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-08-17 22:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2023-08-18 18:57:15.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2023-08-18 18:57:15.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-16 08:31:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 23:49:40.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2023-08-17 13:33:16.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2023-08-18 18:57:15.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"8461538fa57b95837bea89963e1ff8069885d0b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-16 08:19:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5e104c6fb7568a62175da36945fe912824f1f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 08:31:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e73514c726543aab76362da31992752\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb20cdefb6645519b90900fb9ff9f0c : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/586757931bcc4a5cb9ee922aff4bd0d6 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2434664f6beb4c779147290adfaade8d : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/111a50567a154727a0d872a24c0533db : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91eb43a8b5484261b22164fd3d17ed10 : SUCCESS in 3m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2711b08bf24f49d292fecf64962a594c : SUCCESS in 4m 03s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e271ba6e288d4159b74f4b35b073806e : SUCCESS in 2m 55s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/869742438b4647f3b29ca845a3d17cfd : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3473ecc81ba92aa2c9aee155fd1ec3bd80c917a8","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-16 23:49:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"075f2daa81eee026ac59fbd4fbaa103387a72977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 23:57:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bba5697567b4f579a52256fb2f51d92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1571a74a45ce4e4dbbd69cc89b3b4115 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b89b62f26b14f1887cc867a7e3c027d : FAILURE in 2m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f643f7bbdb264f65b7ba064b80948d98 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dfe1cb2bf724d4fbd1615462f041e7a : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8897a59438124204b4bf48ce8111ccdb : SUCCESS in 4m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2ff0e529e5c74c298ddd14c54d303d8c : SUCCESS in 2m 31s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/3a339ba3da5446a4aa324b2c20a1bdd2 : SUCCESS in 3m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/35630dce5f2f40929b22384e7ce51472 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5ba6e1e0ff30d5cd0116cc9b08eb28c398509a56","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-17 13:32:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0101fcbf5f3b6396ae79bc0f558dd78201338aa4","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-17 13:33:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c4aa2fd653577001f67604f327327e14a42696a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 14:16:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0202df81c964b5e96e678b461a1fb68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a7cfecc1c04c5ea0eecf03ceed434a : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27adf80a8bf84998bd5516d00ebea65d : FAILURE in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a09b1c010dd4215be95e72ee9b5a330 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0297bb2d89314038b90f537c8776e8e5 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36534accf2a94cb19a67a3554cf10634 : SUCCESS in 3m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2e08a27ed9d74e03bf65d2b7d65e0918 : SUCCESS in 3m 10s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a61a55b3a33943209fe2759e5d950bb8 : SUCCESS in 4m 05s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8054c2be0eb2478b969ffcd9a4dd53f3 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5f69e7e1a146632d3530e46f29b2b110c331dc93","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-17 21:43:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3725d94e844c4865d3ad73ce31ba6ea465d74eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 22:11:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/594b6e8ed8c445abb131640b6af349b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7ed8947f504bd28821d827640bb6ed : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc197df2f8194ea3806852c2b3ea4b4c : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07b33cedbba74a4282e644de644ad3f1 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d33b11addb0414caccb2435c019821f : SUCCESS in 4m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8841f4042a704e9eb989a4313132420d : SUCCESS in 3m 05s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/ef796c4caa324db9bcff20036b96e576 : SUCCESS in 2m 52s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/278d7f777f1a47edbe00266f78407fb9 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c26221070ac40ef0ec187001b5daaef204e9faf6","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-17 22:33:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a7ed9ba94f372637c3939162371ab9c48e5e68e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-18 18:57:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7bf24ffffe3daec56450f959c7a067ac48f7eb2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 18:57:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8174d45f63fc82eb368ff29c1d708dd5c6e3869","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 19:10:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dee1b93578a241d38ac554f3f27f3226\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d787cab06cba4d82aa28ac5a2cdcfe42 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67d44e5004f04e7eaf66e65600814375 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/707d836c36974f66b5d87b6d2a382cc9 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ca8afe1a6b74461a15c603f391f21d0 : SUCCESS in 3m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ac6071b47c7e44e7b7915efa6060ac57 : SUCCESS in 3m 00s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/d3376daecfaa4716b97e2858e3ff0741 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ba3c1c3d6acecb8d88b0c6b323e5183e498e60bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 19:10:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aed47e0ce613585c107964d5406ed58d7c014647","revisions":{"aed47e0ce613585c107964d5406ed58d7c014647":{"kind":"REWORK","_number":1,"created":"2023-08-16 08:19:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/68/891568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/891568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/891568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/891568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/891568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/891568/1"}}},"commit":{"parents":[{"commit":"afbb77a435d443206505e4577d4452799e163832","subject":"Ignore tox4 .eggs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/afbb77a435d443206505e4577d4452799e163832"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-16 08:19:25.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-08-16 08:19:39.000000000","tz":120},"subject":"[2024.1] Add mnasiadka candidacy for Kolla","message":"[2024.1] Add mnasiadka candidacy for Kolla\n\nChange-Id: I9d550b25ee46d9024f07095cc81d5eae8dcb3842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/aed47e0ce613585c107964d5406ed58d7c014647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/aed47e0ce613585c107964d5406ed58d7c014647"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
