)]}'
{"id":"openstack%2Felection~891611","triplet_id":"openstack%2Felection~master~I9da079c922c62f67aa92cebb0290667654d08adc","project":"openstack/election","branch":"master","topic":"ramereth-2024.1","attention_set":{},"removed_from_attention_set":{"21961":{"account":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"last_update":"2023-09-21 23:57:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9da079c922c62f67aa92cebb0290667654d08adc","subject":"Add Lance Albertson candidancy for Openstack Chef PTL","status":"ABANDONED","created":"2023-08-16 16:16:34.000000000","updated":"2023-09-26 19:17:35.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9a672573227d18e2ffc2483a8568308a56274d2d","_number":891611,"virtual_id_number":891611,"owner":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-16 16:27:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":-1,"date":"2023-08-17 13:40:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-16 16:27:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 23:50:11.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2023-08-17 13:40:22.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2023-08-18 19:08:47.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2023-08-20 06:54:59.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"f3bb8f0e7a80005b5ca0f9a7bf0c1452056ec5f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2023-08-16 16:16:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e028f9c83cf2ba8f91914451d036f254485ae0e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 16:27:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07ad8fc7ba4c4f2ba3c0fdbbe3b1cfd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b0a05967ef4d90a660a18cde1bf29f : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09360408b3b4db4a33fb8c0f4ed4577 : FAILURE in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a2d4f094e374d94b2c79de68ff2d24a : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/132298782a504dfbb31096de3e84aed9 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/447650b2e8654e7591ca7a381afc9b09 : SUCCESS in 3m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0478720464f5438387fceb05ab3506e0 : SUCCESS in 2m 57s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/629bd1f62e4f4693ac8c3e9062b934dd : SUCCESS in 3m 26s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/558ef98bd0ad40d3abebc8609b1723d9 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"297ab06fcfc8e096905dc51749c6928a6623e544","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-16 23:50:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf4cb9bb9e156804f8dd556d888477c4128b1a8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 00:02:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd68151a8e9a4c80ada8aaa74d43c0af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee1e117caa2465c9b1516cea7abf7e9 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18920a070a6474a8d09e9d4e51dc84d : FAILURE in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e66c399e1d64433b81ff20ae2f88cac : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8843509cc92f4ac08c7e9816422a15bf : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93448ab67d134d73abfff771de4594d9 : SUCCESS in 3m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c0c5885842854953866dfb8cdf4bcdf0 : SUCCESS in 3m 32s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d40d3f5369c74310a8003e021fab161a : FAILURE in 3m 01s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/48c9fc05c257487aae698d8c5d9799a8 : FAILURE in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"41c154addfd08643743c1c04c21f7631b2a7b117","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-17 13:39:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d76b2d11d2b2a0ce0d67e08504248d07efc014f7","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-17 13:40:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da28ef8e75b14522067176c48599ba8c82190196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 14:31:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2486984ef8774df6b57039b8c49e336d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34766fc56d51465fa3fa1f0ed01703c6 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc21ad3d4044f0cb4063618ea6f39ac : FAILURE in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/698a2321cb154a5e85a55f4086aeb42f : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4034be0867f4f98a769b9205b55a5d1 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57a31d05f9134f49b30599d2119ef0b3 : SUCCESS in 3m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8cb3ccc58f624e9ca568d560e6683df1 : SUCCESS in 2m 55s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/d36f58d252da4e1b852158f99df7b6ad : FAILURE in 3m 08s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/0c4b997f0a5f4a4a8814ada34918b5e6 : FAILURE in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"64416a5a50b31d5a2aea2a967d44e201a64b8fe9","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-18 19:08:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24b9cf24337bbfea2be9ca7390f26e42b62d77cf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-08-20 06:54:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa6878a6e84d29e11ed54b2220e4bf681fb13083","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-09-21 23:57:13.000000000","message":"Abandoned\n\nThe 2024.1/Caracal is now closed.  This candidacy wasn\u0027t accepted.  Please reach out on #openstack-election , #openstack-tc or via direct email if you\u0027d like to follow-up.","accounts_in_message":[],"_revision_number":1},{"id":"9a672573227d18e2ffc2483a8568308a56274d2d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-09-26 19:17:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"638779c625af31eb92b37c06fa3fc5ebe727e229","revisions":{"638779c625af31eb92b37c06fa3fc5ebe727e229":{"kind":"REWORK","_number":1,"created":"2023-08-16 16:16:34.000000000","uploader":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"ref":"refs/changes/11/891611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/11/891611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/11/891611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/11/891611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/11/891611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/11/891611/1"}}},"commit":{"parents":[{"commit":"afbb77a435d443206505e4577d4452799e163832","subject":"Ignore tox4 .eggs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/afbb77a435d443206505e4577d4452799e163832"}]}],"author":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2023-08-16 16:16:19.000000000","tz":-420},"committer":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2023-08-16 16:16:19.000000000","tz":-420},"subject":"Add Lance Albertson candidancy for Openstack Chef PTL","message":"Add Lance Albertson candidancy for Openstack Chef PTL\n\nChange-Id: I9da079c922c62f67aa92cebb0290667654d08adc\nSigned-off-by: Lance Albertson \u003clance@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/638779c625af31eb92b37c06fa3fc5ebe727e229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/638779c625af31eb92b37c06fa3fc5ebe727e229"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
