)]}'
{"id":"openstack%2Felection~891868","triplet_id":"openstack%2Felection~master~I933fcda89472f1675cc1bf0d37d0e2f9e262cb1f","project":"openstack/election","branch":"master","topic":"charms-2024.1","attention_set":{},"removed_from_attention_set":{"2424":{"account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"last_update":"2023-08-21 01:44:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I933fcda89472f1675cc1bf0d37d0e2f9e262cb1f","subject":"Add Felipe Reyes candidacy for OpenStack Charms PTL","status":"MERGED","created":"2023-08-17 16:22:21.000000000","updated":"2023-08-21 01:45:10.000000000","submitted":"2023-08-21 01:44:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891868-charms-2024.1","meta_rev_id":"261c7441453ea5298760dc9d9270353d064d3754","_number":891868,"virtual_id_number":891868,"owner":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-21 01:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-08-17 21:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2023-08-21 01:33:01.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2023-08-21 01:33:01.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 18:49:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-17 21:44:55.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2023-08-18 18:59:49.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2023-08-21 01:33:01.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"252ff5dcb6c9cf04a6c39b7661be7c1d1fae1495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-08-17 16:22:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"741f11a0ebd824989b8764a8f39a5b93332b2378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-08-17 16:33:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"17e5bcf92f790965526acd7c82fc474c40e0ec8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 18:49:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fcbd69a97504532b22fe1c4ef675d83\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbe0b78d5b5446b83dba38532c493b3 : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0627b9366a354976ac9a15faa65788d5 : FAILURE in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4197965f388e4154944fe54107a01968 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7392724af18147848dbb4d3fe5212b89 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0050a2cc46a14c539e93c5e8f849cb26 : SUCCESS in 4m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/920b74104a6d44db8bbe969bb69716f4 : SUCCESS in 3m 53s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/57a2be72124e4a18b421839ceebf2e2b : FAILURE in 3m 06s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/71cee77c85304e6d88557e43a48816a7 : FAILURE in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"659c6fee27ad444e188bad427044d5f38e62c211","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-08-17 19:34:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7863c7a65f84b2020b7a815e4d377bcf2dbf94bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 19:49:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d52de6bbf594876ba46415e465a9410\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f90abe006da44a3bbfbdfe809493aeb : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/418f5497ca524c6485278f9a760af029 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6025ffc5888246e49ccb83b510c8ba1a : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e95bd817f1c4dbb8f7d74fe43f85aac : SUCCESS in 3m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b8174a2273274de582471abedfa505c6 : SUCCESS in 3m 01s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/f9c15f9c29fe4c9ea8003f20fe71b783 : SUCCESS in 3m 12s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/075179c421d440c8816f7638a74dfb29 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3796190c3ace36fb10b04224839110825a59fb37","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-17 21:44:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd993087367d8b4a850914fdf9e2b5dbd045726c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-18 18:59:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5fabdbbed73009ec2e26acc26a7b1e2b0a60f377","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-21 01:33:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"805b6e6fbfdf83223830abe0a4eb8d9f02cbfa11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 01:33:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"facb063f9d7dc0db50551091151506c77da142c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 01:44:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02e554237b2148d282cd3873084de3b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8337303052b84e77802ac31b65482af6 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f2723799ffd48f8ac978e32cc4984f4 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dc5f5ea284e48e19adf58ce88e1546f : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7aa45b655dbb49b78e5a49dda5539225 : SUCCESS in 3m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/91cf306c6dd14c368538bba1382868ab : SUCCESS in 2m 58s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/7151e50b3c1f4647bb61ca5c059a9770 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6f69646f0ffa1e3cd3d8fc1e88379120fce19cba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 01:44:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"261c7441453ea5298760dc9d9270353d064d3754","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 01:45:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b311eeef68449b3879d23d81316ed8d\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/f5aa399c973e4182b6346d1fc860e2c4 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a28fd9c56c8b3746e0c04e9c7475a81699819029","revisions":{"1ac60ed746136aef9b56f47128dbe906b6a0a3ee":{"kind":"REWORK","_number":1,"created":"2023-08-17 16:22:21.000000000","uploader":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"ref":"refs/changes/68/891868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/891868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/891868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/891868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/891868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/891868/1"}}},"commit":{"parents":[{"commit":"afbb77a435d443206505e4577d4452799e163832","subject":"Ignore tox4 .eggs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/afbb77a435d443206505e4577d4452799e163832"}]}],"author":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2023-08-17 16:21:41.000000000","tz":-240},"committer":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2023-08-17 16:22:19.000000000","tz":-240},"subject":"Add Felipe Reyes candidacy for OpenStack Charms","message":"Add Felipe Reyes candidacy for OpenStack Charms\n\nChange-Id: I933fcda89472f1675cc1bf0d37d0e2f9e262cb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1ac60ed746136aef9b56f47128dbe906b6a0a3ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1ac60ed746136aef9b56f47128dbe906b6a0a3ee"}]},"branch":"refs/heads/master"},"a28fd9c56c8b3746e0c04e9c7475a81699819029":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-17 16:33:01.000000000","uploader":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"ref":"refs/changes/68/891868/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/68/891868/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/68/891868/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/68/891868/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/68/891868/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/68/891868/2"}}},"commit":{"parents":[{"commit":"afbb77a435d443206505e4577d4452799e163832","subject":"Ignore tox4 .eggs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/afbb77a435d443206505e4577d4452799e163832"}]}],"author":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2023-08-17 16:21:41.000000000","tz":-240},"committer":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2023-08-17 16:33:01.000000000","tz":0},"subject":"Add Felipe Reyes candidacy for OpenStack Charms PTL","message":"Add Felipe Reyes candidacy for OpenStack Charms PTL\n\nChange-Id: I933fcda89472f1675cc1bf0d37d0e2f9e262cb1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a28fd9c56c8b3746e0c04e9c7475a81699819029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a28fd9c56c8b3746e0c04e9c7475a81699819029"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
