)]}'
{"id":"openstack%2Felection~892600","triplet_id":"openstack%2Felection~master~I067a8fdb86b211b1f1347259a5a3eb5ee8bcc33e","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"21387":{"account":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"last_update":"2023-08-28 16:56:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I067a8fdb86b211b1f1347259a5a3eb5ee8bcc33e","subject":"Add Cong Ge\u0027s candidacy for Freezer PTL","status":"MERGED","created":"2023-08-25 00:50:10.000000000","updated":"2023-08-28 16:57:53.000000000","submitted":"2023-08-28 16:56:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892600","meta_rev_id":"5a3316e6e3feb3f3ef20cedad825e4cb3edc131f","_number":892600,"virtual_id_number":892600,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 16:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-08-25 12:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2023-08-28 16:39:05.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2023-08-28 16:39:05.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-25 01:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-25 12:25:23.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2023-08-28 16:39:05.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"00567518d3eea1bc9ab0cfdc475d2fbe11481ee6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2023-08-25 00:50:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62e5e6790b1ebca1d94b226933dcb0756b7dc37f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2023-08-25 00:54:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5f8959e43f302776b7ece6ec5b1d75f0fffac598","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2023-08-25 00:54:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"34d7cdea8f770a72b5140a61e06cd6fed84699d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 01:03:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4973f69dff7848b79bc3f40586582511\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a88d3b433b0142248bad8dec1dbe36b0 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7864d19cc53846278a9ac5138c438fbc : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf4733c9e7214c74a3d5a63d8bb7d7f3 : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7efd72e0fa0f4bc791c9b18c8b121940 : SUCCESS in 3m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/124ee1c1e85044b68d9f33cee01a7955 : SUCCESS in 4m 53s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6060569b4923448f9885776c8aa6cbb2 : SUCCESS in 4m 50s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5c783a9b966f44d78db78bab05336863 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"7fc9e817be1513103c50c2a4acc0d68907026617","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-25 12:25:23.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c67646627e40b0dab619be2ec5934ef04fbbeced","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-08-28 16:39:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af92cee14fab8d95666fab902ed98b5fc5ef3010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:39:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c5c2a68c86a14594c0994b82b515ad21d99f88e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:56:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e0de9a706b4606826a9cad1a374c51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de977cc1aee04bf5b2763f88006dbcb8 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c4545a7d1b948a99cbc4cd1e51bde7a : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63c100b097f24acd9e625cf3625a9c63 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cfa97bfc6dd43dda13bc64b5ad55b6f : SUCCESS in 3m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fdff053cd8b14ab09cc37b2a7de90380 : SUCCESS in 4m 32s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/82bd9b9ee7074a139e1a1c8259a0ecd3 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"522fff1845f32e71b47930e0f0f0006f4671a9ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:56:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5a3316e6e3feb3f3ef20cedad825e4cb3edc131f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:57:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/153985d75ede43b0b8ade74c3e0d2b05\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/1b10e9a5c1514214abe4579f3848c6df : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3445727f47e98fc50b66d6308f98838edc635b41","revisions":{"fcc510079b4d8121befde8ae43acdcac2d386d32":{"kind":"REWORK","_number":1,"created":"2023-08-25 00:50:10.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/00/892600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/00/892600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/00/892600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/00/892600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/00/892600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/00/892600/1"}}},"commit":{"parents":[{"commit":"5684fecad1097330903154b7f3245f4bb00cd2f2","subject":"Merge \"Add Martin Kopec candidacy for QA 2024.1 PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5684fecad1097330903154b7f3245f4bb00cd2f2"}]}],"author":{"name":"gecong","email":"ge.cong@zte.com.cn","date":"2023-08-25 00:50:10.000000000","tz":0},"committer":{"name":"gecong","email":"ge.cong@zte.com.cn","date":"2023-08-25 00:50:10.000000000","tz":0},"subject":"Add Cong Ge\u0027s candidacy for Freezer PTL","message":"Add Cong Ge\u0027s candidacy for Freezer PTL\n\nChange-Id: I067a8fdb86b211b1f1347259a5a3eb5ee8bcc33e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fcc510079b4d8121befde8ae43acdcac2d386d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/fcc510079b4d8121befde8ae43acdcac2d386d32"}]},"branch":"refs/heads/master"},"3445727f47e98fc50b66d6308f98838edc635b41":{"kind":"REWORK","_number":2,"created":"2023-08-25 00:54:18.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/00/892600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/00/892600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/00/892600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/00/892600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/00/892600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/00/892600/2"}}},"commit":{"parents":[{"commit":"5684fecad1097330903154b7f3245f4bb00cd2f2","subject":"Merge \"Add Martin Kopec candidacy for QA 2024.1 PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/5684fecad1097330903154b7f3245f4bb00cd2f2"}]}],"author":{"name":"gecong","email":"ge.cong@zte.com.cn","date":"2023-08-25 00:50:10.000000000","tz":0},"committer":{"name":"gecong","email":"ge.cong@zte.com.cn","date":"2023-08-25 00:54:18.000000000","tz":0},"subject":"Add Cong Ge\u0027s candidacy for Freezer PTL","message":"Add Cong Ge\u0027s candidacy for Freezer PTL\n\nChange-Id: I067a8fdb86b211b1f1347259a5a3eb5ee8bcc33e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3445727f47e98fc50b66d6308f98838edc635b41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3445727f47e98fc50b66d6308f98838edc635b41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
