)]}'
{"id":"openstack%2Felection~893297","triplet_id":"openstack%2Felection~master~Icb8c3ff0772b541806371fce058c6e23f3167c66","project":"openstack/election","branch":"master","topic":"template-fix","hashtags":[],"change_id":"Icb8c3ff0772b541806371fce058c6e23f3167c66","subject":"[WIP] Fix on combined election nomination end template","status":"ABANDONED","created":"2023-08-31 00:45:02.000000000","updated":"2024-07-31 16:14:57.000000000","total_comment_count":6,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"440a5ab5c1c2b84514c9f36850fbbcafd0b1b635","_number":893297,"virtual_id_number":893297,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-31 00:53:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 00:53:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6f0782ce90dd4822131aa70123bd8515eaaa7b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-31 00:45:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"914875c8ba49a9a8c52058992a775e26904094e3","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-08-31 00:48:13.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"afba496b946f84616b9fe2fec7cf0e05c81698d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 00:53:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed931c1f644468a8c8c9b7f167dc381\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcdcecbf2eb443b59272d559951d9da8 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55e99f4ab84a410988bfe6faa773eb5f : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78ac1efe05814ba8bd5ce27504e463d9 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8ed08609e234a96a97585d65156373c : SUCCESS in 2m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ac39b4de74946569b01eff43881094d : SUCCESS in 2m 46s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/0a632ec61fc3488ca7680d6f81ccb287 : SUCCESS in 2m 37s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9d8a326b61574d3e80eaa8f7c0bd8939 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"953093d6447acd2646ea683d6f7db77612e46bf7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-07-31 16:13:13.000000000","message":"Abandoned\n\nAddressed through https://review.opendev.org/c/openstack/election/+/910564 instead","accounts_in_message":[],"_revision_number":1},{"id":"440a5ab5c1c2b84514c9f36850fbbcafd0b1b635","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-07-31 16:14:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88b80365bc1fd728f1232ca3ab7b2a124d973617","revisions":{"88b80365bc1fd728f1232ca3ab7b2a124d973617":{"kind":"REWORK","_number":1,"created":"2023-08-31 00:45:02.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/97/893297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/97/893297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/97/893297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/97/893297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/97/893297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/97/893297/1"}}},"commit":{"parents":[{"commit":"8d81d8e4dfb237696df3b04d2d7b6eabb5db5126","subject":"Merge \"Tim Burke candidacy for Swift PTL (Caracal)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8d81d8e4dfb237696df3b04d2d7b6eabb5db5126"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-08-31 00:44:19.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-08-31 00:44:19.000000000","tz":540},"subject":"[WIP] Fix on combined election nomination end template","message":"[WIP] Fix on combined election nomination end template\n\nChange-Id: Icb8c3ff0772b541806371fce058c6e23f3167c66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/88b80365bc1fd728f1232ca3ab7b2a124d973617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/88b80365bc1fd728f1232ca3ab7b2a124d973617"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
