)]}'
{"id":"openstack%2Felection~906912","triplet_id":"openstack%2Felection~master~If6260e6a44fd66ab5df8aa207cb8f4f4bb6ef220","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-02-24 04:24:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If6260e6a44fd66ab5df8aa207cb8f4f4bb6ef220","subject":"Get historical affiliations from the summit API","status":"MERGED","created":"2024-01-26 18:45:09.000000000","updated":"2024-02-24 04:25:03.000000000","submitted":"2024-02-24 04:24:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906912","meta_rev_id":"288163819fe89056dbeeda419ed6b8f053b09e9d","_number":906912,"virtual_id_number":906912,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-24 04:24:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-02-01 13:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2024-02-24 04:10:01.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2024-02-24 04:10:01.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-26 19:02:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-01 13:11:40.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2024-02-24 04:10:01.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"e1d2502d43cf6e6426707648988a20d9edebab5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-01-26 18:45:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac78052bee6472a5df28320b2731daf26db94720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 19:02:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf04897106d47eaaceda13cfda69629\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d6528567524a1bb9ec4e1e6140f392 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2d8379a62454aadbf0902f8c08e0277 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd4d73540f5c4b158c6b6f5b8274b606 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ce50133f5244908b671f6e78551bf30 : SUCCESS in 3m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d92bfe77e034fc2ba764f65520a2b0d : SUCCESS in 3m 15s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/baa5d9b3b2a84795a587af2c3aa3ec16 : SUCCESS in 2m 49s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ddaef1b522b84c948fdd212fe9954180 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8a4762cbabfbb49908e0240db93155f02f222a65","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-01 13:11:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd0e99d3fa1b9038d4ee8f9ec212aa0b1a9b3580","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-24 04:10:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0eb98683a0208a5193f2d440bdd577c271d6144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:10:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"165cf899aad0ce31da40d50f1cd4ee5ee12c6d6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:24:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d2dcfcfa5af46a59ceb3612e7f67392\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124c2584d6f54fef97de0a2a6132feb2 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c57ea04797645fb99c7593d0a86b363 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44cf92f1dd924b24a66f7c590bb13659 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5da8e5115e4f4d2f8b7a0088ca767822 : SUCCESS in 2m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/413d72afc8a3436784fb075ae52c90a0 : SUCCESS in 3m 53s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/695a0fa332544109bb0f09d071a4d076 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a5ddf9398eb1048972ee2b0f370ecd78dda453ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:24:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"288163819fe89056dbeeda419ed6b8f053b09e9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:25:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/283b8377e8f04ad5995d7132b9b08195\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/d24dcb82455746928470a213823c9392 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15a6518a783cbd4df29cd76a5b553a42fd9caae9","revisions":{"15a6518a783cbd4df29cd76a5b553a42fd9caae9":{"kind":"REWORK","_number":1,"created":"2024-01-26 18:45:09.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/12/906912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/12/906912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/12/906912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/12/906912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/12/906912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/12/906912/1"}}},"commit":{"parents":[{"commit":"39a160dc931ca66a84c0e3bc6a1fc88f8fb91b7d","subject":"Merge \"[tox] Exclude the build directory from flake8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/39a160dc931ca66a84c0e3bc6a1fc88f8fb91b7d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-01-26 18:37:20.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-01-26 18:43:39.000000000","tz":0},"subject":"Get historical affiliations from the summit API","message":"Get historical affiliations from the summit API\n\nDuring the transition from OpenStackID to OpenInfraID, some aspects\nof the backend \"summit\" API changed. Historical affiliations ceased\nto be reflected in the \"affiliations\" array, and now need to be\nextracted from \"all_affiliations\" instead. Further, organization\nnames are not included unless \"all_affiliations.organization\" is\nexpanded.\n\nChange-Id: If6260e6a44fd66ab5df8aa207cb8f4f4bb6ef220\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/15a6518a783cbd4df29cd76a5b553a42fd9caae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/15a6518a783cbd4df29cd76a5b553a42fd9caae9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
