)]}'
{"id":"openstack%2Felection~909028","triplet_id":"openstack%2Felection~master~I5ad83be7394f824f02576844b95b850c65d49be1","project":"openstack/election","branch":"master","topic":"2024.2/template","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2024-02-24 04:24:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ad83be7394f824f02576844b95b850c65d49be1","subject":"[trivial] fix minor typo on election_season template","status":"MERGED","created":"2024-02-14 17:21:45.000000000","updated":"2024-02-24 04:25:45.000000000","submitted":"2024-02-24 04:24:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909028-2024.2/template","meta_rev_id":"19f88d832df93a80af357fa748589cd744ccb865","_number":909028,"virtual_id_number":909028,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-24 04:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-02-24 04:11:24.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-02-14 17:22:31.000000000","_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2024-02-14 17:22:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-02-24 04:11:24.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-14 17:22:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-02-14 17:22:31.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-02-14 17:29:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-24 04:11:24.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"4fb4acadeab6f1ed0afdff9896ebd4524c734e50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-14 17:21:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0db81518d2648805dd1c99515bfa40a78061d598","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-14 17:22:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db4815bad4fd6dbaa1c6d189facd72f1e961601d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-02-14 17:22:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15057dc268ce99b4790b0ea340771c6c9a9a6b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 17:29:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f4627ba46894ce6bfb9d73f8b834290\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6393fd5c279499a95cbad72830567d7 : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9214e633ba0a410e9ba65fae2bddbdfb : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea40e489b9ca4c6b978563d8b953e5fd : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d524d309054d4fd58c95d67f4fb6ec5b : SUCCESS in 2m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cd6ee027266c411287c694f53ed7c1ac : SUCCESS in 2m 46s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/3b05d25a7d45409b909b5c575419ea06 : SUCCESS in 3m 01s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a963c0d966a54a77aa1613cf6b907ad8 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c7d0caf759a88a364a731842ea2eab59ef98740e","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-24 04:11:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"020c81fcc3c8e80c57b2bbdad7b7c6e0990825ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:11:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e640ac92558a9a9f0151a28771512d8fbda17425","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:24:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3aef9ff623b449093d1ec317ca7f5d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f8ae769e56940619a7eeae8fceddb78 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d19e5fca604c4d4d9f646183f95faf0e : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89d8670a59d74e0b8544d5fc793762d6 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d24f8f04dbe4e3cb2ecc5ae689ce640 : SUCCESS in 2m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/900f49dee7974acfbf0eb509c974dac7 : SUCCESS in 2m 38s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/dfc85cf292724a238756102422b91af4 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3fe410fb7d2273093e6cf82a0ed7e3ce776ece32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:24:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"19f88d832df93a80af357fa748589cd744ccb865","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 04:25:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f99b800c1024c7d9c87cf5f91e5d20a\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/c577710dbcb344d18d7d6ff588894781 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9846a714b41b618b8f81647b814b37bd4995eda","revisions":{"a9846a714b41b618b8f81647b814b37bd4995eda":{"kind":"REWORK","_number":1,"created":"2024-02-14 17:21:45.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/28/909028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/28/909028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/28/909028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/28/909028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/28/909028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/28/909028/1"}}},"commit":{"parents":[{"commit":"dacf78b1136a19d6826aefb4d4c0944c433e9c64","subject":"[trivial] Remove addition whitespace in string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dacf78b1136a19d6826aefb4d4c0944c433e9c64"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-02-14 17:21:18.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-02-14 17:21:18.000000000","tz":540},"subject":"[trivial] fix minor typo on election_season template","message":"[trivial] fix minor typo on election_season template\n\nChange-Id: I5ad83be7394f824f02576844b95b850c65d49be1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a9846a714b41b618b8f81647b814b37bd4995eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a9846a714b41b618b8f81647b814b37bd4995eda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
