)]}'
{"id":"openstack%2Felection~909313","triplet_id":"openstack%2Felection~master~I179b77c21e5bb2943d70d6ad9ec76e4ba4c811ce","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2024-02-27 22:41:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I179b77c21e5bb2943d70d6ad9ec76e4ba4c811ce","subject":"Adding songwenping candidacy for Cyborg","status":"MERGED","created":"2024-02-18 07:05:03.000000000","updated":"2024-02-27 22:41:28.000000000","submitted":"2024-02-27 22:41:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909313","meta_rev_id":"0047192f0c2636324fe19b5ef88abf23c2da50ad","_number":909313,"virtual_id_number":909313,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-27 22:41:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-02-18 21:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2024-02-27 19:57:07.000000000","_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2024-02-27 19:14:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2024-02-27 22:09:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"date":"2024-02-27 19:57:52.000000000","_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-18 07:13:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-18 21:16:43.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2024-02-27 19:14:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-02-27 19:57:07.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"5a9835759b4a4bddfe69dce7f68f106a521a9c0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-02-18 07:05:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bb220115faa30c42b13d85a48caf05d8e3b6687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 07:13:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68ab4bbe994f478fad55075d936cba20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c63f9f168f25423aa462cf4a111e7c11 : FAILURE in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e6b717b9afc4c96b77bfe0bac7433ff : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17f5687a02064ddf9c45fa8d18643e54 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f323345896e48a8b20a135c9cb167c7 : SUCCESS in 2m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bdb10698165f4d3f848e008c56fac90c : SUCCESS in 2m 30s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/3e8b59a02f444129b8c0e279426a7c83 : FAILURE in 2m 30s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/2b3a9f6f90c5468b9b6da3616ee26286 : FAILURE in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8a381b96024611a8b83d9ca8cc40d373f5501963","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-02-18 09:32:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e81b2629323e83d83bd60fabbda2c2dd6ce6d48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 09:40:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/945526ce65334325af48b4211a4b714b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a0b436e601a4d5d9dcdd2459f1c31f2 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47b19410e22f47639666faa513f4ea7a : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97daa993a3aa4a6491e288a5dc705a5c : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddfcc893acce47b9bffb5457810b9cba : SUCCESS in 2m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/226584f42be94eb5970b2a0e5fe71880 : SUCCESS in 2m 38s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/e4e700d89790461ca842d83b66cb19c0 : SUCCESS in 2m 38s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a708d316976244619c3638aab0228f62 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e14db64037179fdcca6ff6f85f8bfb958b269e30","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-18 21:16:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3e5466a7f06adc575cadb2fa38fdb86b046d3f5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-27 19:14:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a61ee68e9b18fb6446d07df8c80b1c6662d83b69","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-02-27 19:57:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b6d9b96d1e4303d1bf2556596a232debd6a45bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 19:57:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3616833530c168a21709e93886acce2fa75c4b2","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-02-27 19:57:52.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69308adeb095101685f22ef243ef0d8b570ddd11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 19:58:04.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"44d2e2d1df4b89bd70db310cdc1a4767ff1913ba","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-27 22:09:54.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3da1387a42c1f22a6f7a2afef3c01e668ca64e03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:30:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/390662da15f34c3484958c5f8da92477\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97910fca2c124be494aa5ef8beebed62 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4c2b817354245aab84814afa6fd77a1 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6c0f3d053d246059c4cb8d41122f783 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f023436b4d684f58a4910389af70d708 : SUCCESS in 2m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e004767e9971444db2c5d79b3ba230b2 : SUCCESS in 2m 49s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/1f987ba549a842edb400f5c13820d169 : SUCCESS in 4m 23s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/323bd046f3bd45d4b3ec494833a61684 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5e33c381f97140077b707c1073784ee1cd5197e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:31:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76f2a03dcbe6eaef0240c299aff49d75ceaa6019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:41:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1be05ed1c23645ae976eeabdc8080317\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4249aae516b4b72a3aa70366fdc43d9 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4598f1e9afc74a799cb7e60872900896 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d853b1cb8c204a80816c7e9977c097ec : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7be9ffdebb74c769efdf5879ec46c5d : SUCCESS in 3m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/47145c606a7b45bfad267964ab20aef5 : SUCCESS in 2m 52s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9c682bf23a864468980ab992074fcae1 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0047192f0c2636324fe19b5ef88abf23c2da50ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:41:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8db6da1cd0b3d6fa02a83dbbf745e0cf286e832","revisions":{"a8db6da1cd0b3d6fa02a83dbbf745e0cf286e832":{"kind":"REWORK","_number":1,"created":"2024-02-18 07:05:03.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/13/909313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/13/909313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/13/909313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/13/909313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/13/909313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/13/909313/1"}}},"commit":{"parents":[{"commit":"0d8ba9e0f541f2dda71c7d19fefd4fc64a8508ce","subject":"Create candidates/2024.2 placeholder directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0d8ba9e0f541f2dda71c7d19fefd4fc64a8508ce"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2024-02-18 07:02:07.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2024-02-18 07:02:12.000000000","tz":480},"subject":"Adding songwenping candidacy for Cyborg","message":"Adding songwenping candidacy for Cyborg\n\nChange-Id: I179b77c21e5bb2943d70d6ad9ec76e4ba4c811ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a8db6da1cd0b3d6fa02a83dbbf745e0cf286e832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/a8db6da1cd0b3d6fa02a83dbbf745e0cf286e832"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
