)]}'
{"id":"openstack%2Felection~909626","triplet_id":"openstack%2Felection~master~I018172c732fe3c6aa0db1f239a7b5e6ecdfab025","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"18418":{"account":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"last_update":"2024-02-26 00:28:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I018172c732fe3c6aa0db1f239a7b5e6ecdfab025","subject":"Adding Andy Botting candidacy for Murano","status":"ABANDONED","created":"2024-02-21 00:42:25.000000000","updated":"2024-02-26 00:28:34.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a79e6c98e60c989e85df0fffff2fd16c1933f0db","_number":909626,"virtual_id_number":909626,"owner":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-21 00:50:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":-1,"date":"2024-02-21 10:59:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-21 00:50:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-21 10:59:25.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"5edac9136bfcb494dc49a2635b45039d4cb054f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2024-02-21 00:42:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b204ecf99f3eea5c3cd8e818357802e9b5f7a2eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 00:50:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fec8ef347880440f81c861df5f07a071\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a070d3abe8cf4ce98bf170f677859884 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5e25d61a78c48cda577262b4ecfa322 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69224dc4ba39482db7ca1345d5c0c96a : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/296d54b6f8bd43028b8fa78313d4c895 : SUCCESS in 3m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e774835134ea4b498ed4520a7a703f2e : SUCCESS in 2m 34s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/35771392e0b448cdbdf4db92ee647af7 : FAILURE in 3m 08s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b673b878f9fb472697f9bd7c99536f15 : FAILURE in 2m 29s","accounts_in_message":[],"_revision_number":1},{"id":"8a34573ecd559ab6d30d2529b935b583ee3f0823","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-21 10:59:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a79e6c98e60c989e85df0fffff2fd16c1933f0db","tag":"autogenerated:gerrit:abandon","author":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"date":"2024-02-26 00:28:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dba7506bde5f9d69bbea6d63e30c9ba924f2f201","revisions":{"dba7506bde5f9d69bbea6d63e30c9ba924f2f201":{"kind":"REWORK","_number":1,"created":"2024-02-21 00:42:25.000000000","uploader":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"ref":"refs/changes/26/909626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/26/909626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/26/909626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/26/909626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/26/909626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/26/909626/1"}}},"commit":{"parents":[{"commit":"0d8ba9e0f541f2dda71c7d19fefd4fc64a8508ce","subject":"Create candidates/2024.2 placeholder directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0d8ba9e0f541f2dda71c7d19fefd4fc64a8508ce"}]}],"author":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2024-02-21 00:41:18.000000000","tz":660},"committer":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2024-02-21 00:41:18.000000000","tz":660},"subject":"Adding Andy Botting candidacy for Murano","message":"Adding Andy Botting candidacy for Murano\n\nChange-Id: I018172c732fe3c6aa0db1f239a7b5e6ecdfab025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dba7506bde5f9d69bbea6d63e30c9ba924f2f201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dba7506bde5f9d69bbea6d63e30c9ba924f2f201"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
