)]}'
{"id":"openstack%2Felection~910024","triplet_id":"openstack%2Felection~master~I4c44cda711b30be15615bd7655b7780df84d57fe","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-02-27 22:40:13.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-02-27 22:40:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c44cda711b30be15615bd7655b7780df84d57fe","subject":"Adding Jon Bernard candidacy for Cinder PTL","status":"MERGED","created":"2024-02-23 19:28:00.000000000","updated":"2024-02-27 22:41:21.000000000","submitted":"2024-02-27 22:40:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910024","meta_rev_id":"8bb6516739fa1107fa6ba467e3fbabdfdee211ea","_number":910024,"virtual_id_number":910024,"owner":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-27 22:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-02-26 13:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2024-02-27 19:32:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-26 16:33:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2024-02-27 22:08:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-23 19:40:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 19:48:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2024-02-26 13:03:45.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2024-02-26 16:33:35.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-02-27 19:32:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-02-27 19:46:47.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"}],"messages":[{"id":"de4f374f7a6eb1fe39d78784ff361a4fc5c3f07a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-02-23 19:28:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c589cdde68290fba0ec67a584f8deb13a7af6b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 19:40:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11b12e86d81c4fceb850125c566aa56f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097a779bf0cc44778af5585417e59ad1 : FAILURE in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8794216e857247b88eba57fbe1ba8cec : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/273f974933954c2cb0eb451f02d9ab36 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10a57ad7560f474fa1c373dada21688c : SUCCESS in 4m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b57107d405034147b701f7b7b31375f7 : SUCCESS in 2m 41s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/323ea69bbe0f412c971d5969489c0629 : FAILURE in 3m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ac789d4ed65b4e5ca690e887c70aa882 : FAILURE in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d7849fc0b78ff008a0615a7ccd6ceeb95522d922","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-02-23 19:48:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e6ad96da53c478bd2027a86912273e55665cb39","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-02-23 20:35:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dc7bf3d3c55a48f087a777c04f73c86db1d0934","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-02-23 20:36:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d37212ec87a3235e23925558cf75472d44041af8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 20:43:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cef37ce85ae24d19947215fefc59a028\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971f6f90718b45429510574b57f8a3d7 : FAILURE in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba30a35a55434d79a5255f90fe9bbe24 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/678471c86c6d425a8d3d0a11e5d75f2e : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d592b10a0e2f4816adb666a586229e4b : SUCCESS in 2m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7470e6fd6b1a488484215d703dc2eadc : SUCCESS in 2m 48s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/5ff0d44dcd68499d8377a4c62d2be378 : FAILURE in 2m 49s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a5dbb09e03b24bb4914571fdc123a142 : FAILURE in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"c0e7258719f1f5eaba08a9afdd28d91b9429d536","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-02-23 21:04:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c0e00d1cef08d6ce63caa1202b68f96ef7e1efe","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-02-23 21:11:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7e0f0a6364a217558a73170d84ed388a3cc4947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 21:14:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cff74f4675314d37b0d3c0753e5103f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9515e68a922a4fa9810e5a89c7fc14e3 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2346ead7046641d98f2f0ba9c8385f82 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a61520f3d94c4226815577928341e048 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38fd9ea99ed3467590bfe36843ea554d : SUCCESS in 3m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6b6836d1d9864c86943d35ee8bb8019e : SUCCESS in 2m 55s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/87ecad4e383b4658add4cbd9e1ee5318 : SUCCESS in 2m 57s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3855ae197e9244dba6553cef4639d588 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"592369208d26c290b738a9c9220d97db9884adab","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-26 13:03:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b6699cd650514fa7f058296c5c8f7d70cdab830","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-26 13:04:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4bed283fabe7f66fdfc636bdfe7c69550fe1843","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-26 16:33:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"701c2e42ee9fccd33363237259edccaeb76173cf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-27 19:32:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4be38d3752855f8c298a696016123e5e04595895","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-02-27 19:46:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f37281a6268af66a01349728849a03a90b7e06ab","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-27 22:08:30.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebd8c9b78a9e664705d491336cf0e6349c69e6c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:09:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b4d7e0229ee827d050e18e9b13927cb5b3199d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:40:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26d7f98d96dc4e5391ddee54d812cf03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9589adce054135bbbcc87e2c5a32ce : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/242a2360d4074681a433ea199689b9d3 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a88bc3dbf4a54ef7981cb52d59f2f239 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40aebaed1f704a7590dc2ce3899fa35d : SUCCESS in 3m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/88bd37f5f5cd44e39fe7d60fe9252b5b : SUCCESS in 3m 02s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1be71fac6d234ec992f27624b5e4d510 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4d920882cf26e5603a7ce0fae33fa817f06ee52e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:40:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8bb6516739fa1107fa6ba467e3fbabdfdee211ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 22:41:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05247be1dd36443488c079078b00f308\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/7806da6681a34ce0b076a72e986388ae : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38adba0046c08fd53a729851d6b90b2d1bdd9072","revisions":{"38adba0046c08fd53a729851d6b90b2d1bdd9072":{"kind":"REWORK","_number":1,"created":"2024-02-23 19:28:00.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/24/910024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/24/910024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/24/910024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/24/910024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/24/910024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/24/910024/1"}}},"commit":{"parents":[{"commit":"0d8ba9e0f541f2dda71c7d19fefd4fc64a8508ce","subject":"Create candidates/2024.2 placeholder directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0d8ba9e0f541f2dda71c7d19fefd4fc64a8508ce"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2024-02-18 16:29:34.000000000","tz":-300},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2024-02-23 19:28:00.000000000","tz":-300},"subject":"Adding Jon Bernard candidacy for Cinder PTL","message":"Adding Jon Bernard candidacy for Cinder PTL\n\nChange-Id: I4c44cda711b30be15615bd7655b7780df84d57fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/38adba0046c08fd53a729851d6b90b2d1bdd9072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/38adba0046c08fd53a729851d6b90b2d1bdd9072"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
