)]}'
{"id":"openstack%2Felection~914925","triplet_id":"openstack%2Felection~master~I1289a8bd50d2651340697adac92745e084de8842","project":"openstack/election","branch":"master","topic":"add-tc-resolution","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2024-04-03 22:57:55.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-04-03 22:44:52.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-04-03 22:57:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1289a8bd50d2651340697adac92745e084de8842","subject":"Add 2024.2 Election TC resolution link","status":"MERGED","created":"2024-04-02 20:47:27.000000000","updated":"2024-04-03 22:58:57.000000000","submitted":"2024-04-03 22:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"914925-add-tc-resolution","meta_rev_id":"60b0e9d0951bcd20e3217ef67ef89e7d916ee4fd","_number":914925,"virtual_id_number":914925,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-03 22:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-04-03 22:44:52.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-04-03 22:44:52.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-02 20:49:25.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-04-02 21:09:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-03 01:26:48.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"3162fef45bbf65ad76322d85e25793827f5a0656","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-04-02 20:47:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"014ec8ce94129a4679d423802725583bf5af38c7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-02 20:49:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4281164da19260589e6d331469f2f5e03df29684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-04-02 20:52:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"cc1f8161165e1831c07cb175ecadbd830bd7ee9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-04-02 20:53:02.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e442a301392ac4ca202106eccb17c505339409af","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-02 20:53:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"38359135b9524abf344929a6e9ddfcb522b1982d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-02 21:09:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3d1935e6bd64005a1ac7aebd13bd214\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10963ad9db1461da016edcd3a1098cd : SUCCESS in 2m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a5ac502dd6345ae92f56587d049f563 : SUCCESS in 2m 58s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/489d5cce89be49b79e40442aa1af1b2a : SUCCESS in 2m 51s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c6e0859e887147b385605d4234cbcbf0 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":3},{"id":"5f75a4f50cbf1a9a904ce159bff67c25780a3d9b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-04-03 01:26:48.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"33563adab8106f9dfe31f414b285eca6321105ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-04-03 10:03:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ed9dbd4c753b3c3cb4ac97a9d932e7ace858f929","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-04-03 10:06:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2382653b08a3fd2a2741ab0a41460fc60824e2d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 10:19:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c9b5e5a206e4fa8a72094aec712a004\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2f71e16d88f443cb4ff2cd342ece52f : SUCCESS in 2m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/30bdb546ae644cb2b3c96521f405331e : SUCCESS in 2m 25s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/905782f62d1345bf8a22c7a949a0ba47 : SUCCESS in 2m 24s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a715700e2252445e9bdd258c69d4ab15 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":4},{"id":"8ba76df2f2de568fceda57bdc45cb40a144a3a03","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-04-03 22:44:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"446d14dcdcc30bdb38251584d379a7401fdcefd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 22:45:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"71e84835b55856fafb8970169c6e49f37250d0ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 22:57:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/827087c7bc3040aaafefda48b74475b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/affd41b1814c481396be1a4945c01c60 : SUCCESS in 3m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ecfb3363b3d4a4fb941ca1ce05fc551 : SUCCESS in 5m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/cfbe2988645b4673b907d805d2d5d64c : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":4},{"id":"d44493fbb7deee601e65c0fc609ee90656cf1025","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 22:57:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"60b0e9d0951bcd20e3217ef67ef89e7d916ee4fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 22:58:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a0171423a2a4c439c0fbc123133e93d\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/4389083df0cf44588356aeb5811c7bdc : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"78da72bbf501f394ad33c6843c03a815b6857c29","revisions":{"9e1d1231cec6be0a71de46741ce2345303f59542":{"kind":"REWORK","_number":1,"created":"2024-04-02 20:47:27.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/25/914925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/25/914925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/25/914925/1"}}},"commit":{"parents":[{"commit":"7595ffb42198d35994ad066e2049a9e000865b1b","subject":"Merge \"Update tc-election-summary.py: reflect results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7595ffb42198d35994ad066e2049a9e000865b1b"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-04-02 20:46:36.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-04-02 20:46:36.000000000","tz":540},"subject":"Add 2024.2 Election TC resolution patch link","message":"Add 2024.2 Election TC resolution patch link\n\nChange-Id: I1289a8bd50d2651340697adac92745e084de8842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9e1d1231cec6be0a71de46741ce2345303f59542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9e1d1231cec6be0a71de46741ce2345303f59542"}]},"branch":"refs/heads/master"},"4816cf71630b30891d948e8da024658f591c7d19":{"kind":"REWORK","_number":2,"created":"2024-04-02 20:52:46.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/25/914925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/25/914925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/25/914925/2"}}},"commit":{"parents":[{"commit":"7595ffb42198d35994ad066e2049a9e000865b1b","subject":"Merge \"Update tc-election-summary.py: reflect results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7595ffb42198d35994ad066e2049a9e000865b1b"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-04-02 20:46:36.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-04-02 20:52:36.000000000","tz":540},"subject":"Add 2024.2 Election TC resolution patch link","message":"Add 2024.2 Election TC resolution patch link\n\nChange-Id: I1289a8bd50d2651340697adac92745e084de8842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4816cf71630b30891d948e8da024658f591c7d19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/4816cf71630b30891d948e8da024658f591c7d19"}]},"branch":"refs/heads/master"},"8ee8708d07da0cf3df532b33f7152d256a744cd2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-04-02 20:53:02.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/25/914925/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/25/914925/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/25/914925/3"}}},"commit":{"parents":[{"commit":"7595ffb42198d35994ad066e2049a9e000865b1b","subject":"Merge \"Update tc-election-summary.py: reflect results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7595ffb42198d35994ad066e2049a9e000865b1b"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-04-02 20:46:36.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-04-02 20:52:49.000000000","tz":540},"subject":"Add 2024.2 Election TC resolution link","message":"Add 2024.2 Election TC resolution link\n\nChange-Id: I1289a8bd50d2651340697adac92745e084de8842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8ee8708d07da0cf3df532b33f7152d256a744cd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8ee8708d07da0cf3df532b33f7152d256a744cd2"}]},"branch":"refs/heads/master"},"78da72bbf501f394ad33c6843c03a815b6857c29":{"kind":"REWORK","_number":4,"created":"2024-04-03 10:03:50.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/25/914925/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/25/914925/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/25/914925/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/25/914925/4"}}},"commit":{"parents":[{"commit":"7595ffb42198d35994ad066e2049a9e000865b1b","subject":"Merge \"Update tc-election-summary.py: reflect results\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/7595ffb42198d35994ad066e2049a9e000865b1b"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-04-02 20:46:36.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-04-03 10:02:18.000000000","tz":540},"subject":"Add 2024.2 Election TC resolution link","message":"Add 2024.2 Election TC resolution link\n\nAdding the link as note section with explanation if exists\nas jinja template\n\nChange-Id: I1289a8bd50d2651340697adac92745e084de8842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/78da72bbf501f394ad33c6843c03a815b6857c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/78da72bbf501f394ad33c6843c03a815b6857c29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
