)]}'
{"id":"openstack%2Felection~925927","triplet_id":"openstack%2Felection~master~I7a28e6cec6c46af35ac9bf5ebf30fcec71fbbd43","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-08-18 13:29:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a28e6cec6c46af35ac9bf5ebf30fcec71fbbd43","subject":"Include current affiliation in job output","status":"MERGED","created":"2024-08-08 00:07:11.000000000","updated":"2024-08-18 13:30:12.000000000","submitted":"2024-08-18 13:29:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925927","meta_rev_id":"a1250ed2b8b3403676ee0310d571d96a7649f828","_number":925927,"virtual_id_number":925927,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-18 13:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-08-12 22:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2024-08-18 13:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-08 00:20:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-12 22:10:55.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"c224355d7e5caef434c7bf5bc107335b131732d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-08-08 00:07:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2df2fe93a6b5bb2728952792a33773225c3999e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 00:20:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7e4ceaa019e4825b70d9fc4a6c29d8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb2c2c6adba14169aced7f5da13d502c : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d38d4f7976b44a28b189c5b3043fd1a : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32c15e118e8e4543b05f6e105ee90d79 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e07cc31db644048ba22edadf3f470b2 : SUCCESS in 5m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a09092f162a343898748f1a44602d037 : SUCCESS in 2m 43s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2306080c6de44063b33fd96b64266d61 : SUCCESS in 2m 36s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/02e38c976bcc4ef6bad69cbba601effc : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1},{"id":"142031a7d7069cdb5a5f48067fa2672ab84b2078","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-12 22:10:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"542d7c59aa36198a8d08507a4342cc0a1e014314","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-18 13:18:24.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1b383c74e2dd78ce23bdf0eeba8157c7de35689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-18 13:18:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fb3e9a713f07a975518276f0b5d97fa0f8753b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-18 13:29:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fe5569c962e410c8213bdec30c78523\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fab91968d34947f59dde5beaa3d72cc9 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22c1bbc8e8524a7785bd9b793701b157 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d04c94515d1c4adb9457501c27756a01 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1cc97682072478498f9655168200d17 : SUCCESS in 2m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1e62b0cefaf447a8a4b9e2bb1cb034be : SUCCESS in 2m 33s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/f92ab644787441c6bbb0cdf6ca420617 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a453bb0dccc7b614ddd01a8f2eebee9e63631e10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-18 13:29:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a1250ed2b8b3403676ee0310d571d96a7649f828","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-18 13:30:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4783319053d4207b4b531eca98dd3c4\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/cfab7943744c4187aab7ee31e76938a1 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db3d30a466c6ae4e57b4045fef0e8c85f6cf3af8","revisions":{"db3d30a466c6ae4e57b4045fef0e8c85f6cf3af8":{"kind":"REWORK","_number":1,"created":"2024-08-08 00:07:11.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/27/925927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/27/925927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/27/925927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/27/925927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/27/925927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/27/925927/1"}}},"commit":{"parents":[{"commit":"2647c469ccb79a50b2803a44b1bb1865080d829e","subject":"Merge \"[templates] Update links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2647c469ccb79a50b2803a44b1bb1865080d829e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-08-08 00:04:37.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-08-08 00:04:37.000000000","tz":600},"subject":"Include current affiliation in job output","message":"Include current affiliation in job output\n\nWe can\u0027t fail a nomination based on affiliaton, but it\u0027s somewhat useful\nto see it when evaluation a candidate.  This change displays the current\naffiliation for the candidate under review.\n\nChange-Id: I7a28e6cec6c46af35ac9bf5ebf30fcec71fbbd43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/db3d30a466c6ae4e57b4045fef0e8c85f6cf3af8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/db3d30a466c6ae4e57b4045fef0e8c85f6cf3af8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
