)]}'
{"id":"openstack%2Felection~926177","triplet_id":"openstack%2Felection~master~I61a686a3dc4e18557e5ada1d54db5d80c716a124","project":"openstack/election","branch":"master","topic":"sunbeam-ptl","attention_set":{},"removed_from_attention_set":{"35761":{"account":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"last_update":"2024-08-15 17:24:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I61a686a3dc4e18557e5ada1d54db5d80c716a124","subject":"Add Guillaume Boutry candidacy for Sunbeam 2025.1 PTL","status":"MERGED","created":"2024-08-13 08:36:22.000000000","updated":"2024-08-15 17:25:46.000000000","submitted":"2024-08-15 17:24:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926177-sunbeam-ptl","meta_rev_id":"2c5637091cf43e74ed6b64d0c3b4f22541ce3a3b","_number":926177,"virtual_id_number":926177,"owner":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-15 17:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-08-15 13:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2024-08-15 17:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2024-08-15 17:06:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-13 08:44:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-14 23:58:54.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2024-08-15 13:09:32.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2024-08-15 17:06:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5aa66765ed4a8f52845c831aa974e77455f88e17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2024-08-13 08:36:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a89a30a7cefd35d324b76c67a9471b75dfc9038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 08:44:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5894420862c4be087b89f8089102bd0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f6f38668b94a9a9f2814976ed84021 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62616d7407844fb3aa2ad0ce9dae4aa5 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5f5dcc6c0b043648d528e96596f5f50 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c91d60075e7409595b213f80e4be906 : SUCCESS in 3m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b09b9f4385ea4835bebc90ddad820d1c : SUCCESS in 2m 26s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7cde8a58f2f04d8689e2209ed8acaf5e : SUCCESS in 2m 38s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3a525117fa0f4a5eb115460dc92fdd11 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b6c7c0266df4e4d09632773e8137aff1f70f5415","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-14 23:58:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73f61b9356f98d232a2e4ae7cb79036e3b60be59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 00:10:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7d7dbb6f0bf4668a27e8c64e7099bce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b3fa5f949e419fbab5cbb8e466b977 : FAILURE in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d16f500d233c4e04ba483c990f4854b2 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57234245f3d84a66b109eaa386eecc25 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47f772df65df434ea6f703715700f13a : SUCCESS in 2m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6d6ec006907f41dda61dc325d8e3d682 : SUCCESS in 2m 23s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/ab5735cb2f294e3eac1d39d474d65360 : FAILURE in 3m 37s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9ce72dddc1f74bf8bc67794f6038cb2d : FAILURE in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"86783fbb10087353b3f1f0ecc20506a8b380e300","author":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"date":"2024-08-15 06:40:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"206722e452105ac6774e638f1f1fc9b86b0874a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 06:49:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac70a0f3f4844fb789822136db8d85f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de72a0ca8134091adbb0c9ebbceaeb2 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e04f2cb9dec49309330c7912b320b85 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90908e09b6c348a79ee33380c509558b : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9debaa4263d49669e800b858eb79dee : SUCCESS in 2m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b957518fb27e45b8bc2e17d87897b1a9 : SUCCESS in 2m 41s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8503b74c5a604e9e8297fbf9ca2e1a86 : SUCCESS in 2m 54s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c85e640f888d4935b1d15d3a4575264e : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"78a2c820f8689a40ca5ba783e490b5e1d1c24f86","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-15 13:09:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"181a506ee0e442f39aa58de8a6337a702581ba24","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-15 17:06:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"801f45955519122b6083b818286a67718fd0f41b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:06:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08534e799c32a0261dca519a1dac9617f77aee37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:24:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9c86c04372f4df98c4a6579e9a9542f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/168f5319ab5848b69054de8df9c306a6 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/034e2796765e42459b3f048da827ee9a : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac05c3ed7f0541f38fb02aac891a7daf : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/734d90d6d57f4f1787ebf4524b9f6adf : SUCCESS in 2m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/790a9660f4734bc2b71aebd2ed15574a : SUCCESS in 3m 20s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/bcbc24b7fafa40c8acc5b707f83c0d70 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"92d91123f9e1ba2b8f039284a3b814b94c136ecc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:24:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2c5637091cf43e74ed6b64d0c3b4f22541ce3a3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:25:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/808818d13f7e47b7bd30508cdd8750f0\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/e68a6e61aa854e7eb5151d632b04d541 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5215404308315c435efb3c2ad0c398212f90745","revisions":{"d5215404308315c435efb3c2ad0c398212f90745":{"kind":"REWORK","_number":1,"created":"2024-08-13 08:36:22.000000000","uploader":{"_account_id":35761,"name":"Guillaume Boutry","display_name":"gboutry","email":"guillaume.boutry@canonical.com","username":"gboutry"},"ref":"refs/changes/77/926177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/77/926177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/77/926177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/77/926177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/77/926177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/77/926177/1"}}},"commit":{"parents":[{"commit":"69ffd2437a3020df7d495f6efaa1e30aa230e9de","subject":"Add Watcher candidates/2025.1 placeholder directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/69ffd2437a3020df7d495f6efaa1e30aa230e9de"}]}],"author":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2024-08-13 08:34:01.000000000","tz":120},"committer":{"name":"Guillaume Boutry","email":"guillaume.boutry@canonical.com","date":"2024-08-13 08:34:01.000000000","tz":120},"subject":"Add Guillaume Boutry candidacy for Sunbeam 2025.1 PTL","message":"Add Guillaume Boutry candidacy for Sunbeam 2025.1 PTL\n\nChange-Id: I61a686a3dc4e18557e5ada1d54db5d80c716a124\nSigned-off-by: Guillaume Boutry \u003cguillaume.boutry@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d5215404308315c435efb3c2ad0c398212f90745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/d5215404308315c435efb3c2ad0c398212f90745"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
