)]}'
{"id":"openstack%2Felection~926180","triplet_id":"openstack%2Felection~master~I283ef3fe43b2307a6b16932fcbfc52aae1ed76ec","project":"openstack/election","branch":"master","topic":"ptl/2025.1/openstack-charms","attention_set":{},"removed_from_attention_set":{"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2024-08-15 17:26:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I283ef3fe43b2307a6b16932fcbfc52aae1ed76ec","subject":"Add James Page as candidate for OpenStack Charms PTL","status":"MERGED","created":"2024-08-13 08:49:59.000000000","updated":"2024-08-15 17:26:59.000000000","submitted":"2024-08-15 17:26:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926180-ptl/2025.1/openstack-charms","meta_rev_id":"49722bd8cb8999f752c756f762022f774ceaf3dc","_number":926180,"virtual_id_number":926180,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-15 17:26:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-08-15 13:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2024-08-15 17:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2024-08-15 17:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-13 08:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-14 23:58:37.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2024-08-15 13:07:50.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2024-08-15 17:07:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"6d339c5b80eefc1b0cc709bad06263aa83759211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-08-13 08:49:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79bb498ad213d4d7ccb4ebce12260b561079c064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 08:59:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdd656f3aa2847f1a256adad33111450\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f059818c1624b248c3b45eae08823de : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbcff239009a45c799e6e9b143b41edf : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c158aa3125e440c4a98243deec88a000 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fffc0952a60a49329bacd1bcb5c0b340 : SUCCESS in 2m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9ce12c6601824bdfb511ca6abab8125d : SUCCESS in 2m 42s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/97b4e2263184492da48ab621dc965ca8 : SUCCESS in 2m 46s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b257aa26096b4d6dbb8fae12afdd311a : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f5641dd26e492593fb8dcf97054d3e8d228ade53","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-14 23:58:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0934b5af08265396865990cce95eafa55e45efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 00:14:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff6a45fdd74a4ef9a873bf9cffa81b75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc90af9230ad4084b8fdb116993fa8fb : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd7b240f15914fa4bcd1fc0f19dabd01 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49f70a58c9fd438d99c03eca9ac17efa : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/738e012d227a42dfa7778a113bb008be : SUCCESS in 2m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/28362904d8824b90b50f0d468498d5b7 : SUCCESS in 2m 20s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/2ba1b053ee294ae89339aa75793a268f : FAILURE in 3m 33s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/8b86cef9a54f4e5b87d8ce82df503d20 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cb26eef2ae7507ae482a6ca661c5720617c99a9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-08-15 08:38:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dfee5ef8fc1742de4d1c10b89a86c42819365a85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 08:43:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7e78cf39eda46c8b8cfaba49a8b58f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7039ad7e808f48209d6c491d9a161094 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8374d64b1ac34c4b9088135cd399c47a : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e15b1e062cb54ad1a17d12097abe569d : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3c0b14fc6fd4167a0f1ca82df6e3d3b : SUCCESS in 2m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4fa0aa12268346d6b4bbff5d30917b5c : SUCCESS in 2m 53s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/aeb5ec92ae3e49898c1868eccffb879c : SUCCESS in 2m 43s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/d1e0e411ec7f449d8ea45c254cc7669f : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":2},{"id":"8b02d1010dc57f88ae8726c2372ee5d8dd6981ab","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-15 13:07:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9064e4a66db0bbbac00b68361f80ed2fe81011b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-15 17:07:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"18987280447f56beb44e0101fa4dda39fc79cc40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:08:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d3b2995b6c7289639e02671fc735eafc87b73549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:26:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66c184e376c246a699b9890e32053820\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e418b46e1649c48c2ff4fbb40315c3 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9442ba999ecc4915922de845c62aabf5 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4d4b7e3b5e1482ea2834842cca2eb39 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42566183d77b442190a6c469e9b2abe3 : SUCCESS in 4m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/909d7b7b652a402e8a6e6f867737ead6 : SUCCESS in 3m 44s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b11b9b5dd46e48ff96e7f28d41120b5f : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"2ec3fe8b0cb62577387a983ae52e95c4b0b7720d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:26:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"49722bd8cb8999f752c756f762022f774ceaf3dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 17:26:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a854930db67415fb39b9b130150d164\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/fcaca285c332495ab1a2109f33d79794 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e8ae40a9ce8666596f3c451eb8c2b09a6ff2f13","revisions":{"0f289f42d06764700cf2ee38fec63b979eb2b5b3":{"kind":"REWORK","_number":1,"created":"2024-08-13 08:49:59.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/80/926180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/80/926180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/80/926180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/80/926180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/80/926180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/80/926180/1"}}},"commit":{"parents":[{"commit":"69ffd2437a3020df7d495f6efaa1e30aa230e9de","subject":"Add Watcher candidates/2025.1 placeholder directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/69ffd2437a3020df7d495f6efaa1e30aa230e9de"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-08-13 08:49:14.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-08-13 08:49:35.000000000","tz":60},"subject":"Add James Page as candidate for OpenStack Charms PTL","message":"Add James Page as candidate for OpenStack Charms PTL\n\nChange-Id: I283ef3fe43b2307a6b16932fcbfc52aae1ed76ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0f289f42d06764700cf2ee38fec63b979eb2b5b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/0f289f42d06764700cf2ee38fec63b979eb2b5b3"}]},"branch":"refs/heads/master"},"1e8ae40a9ce8666596f3c451eb8c2b09a6ff2f13":{"kind":"REWORK","_number":2,"created":"2024-08-15 08:38:40.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/80/926180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/80/926180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/80/926180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/80/926180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/80/926180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/80/926180/2"}}},"commit":{"parents":[{"commit":"69ffd2437a3020df7d495f6efaa1e30aa230e9de","subject":"Add Watcher candidates/2025.1 placeholder directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/69ffd2437a3020df7d495f6efaa1e30aa230e9de"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2024-08-13 08:49:14.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2024-08-15 08:38:33.000000000","tz":60},"subject":"Add James Page as candidate for OpenStack Charms PTL","message":"Add James Page as candidate for OpenStack Charms PTL\n\nChange-Id: I283ef3fe43b2307a6b16932fcbfc52aae1ed76ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1e8ae40a9ce8666596f3c451eb8c2b09a6ff2f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1e8ae40a9ce8666596f3c451eb8c2b09a6ff2f13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
