)]}'
{"id":"openstack%2Felection~926581","triplet_id":"openstack%2Felection~master~I5757771fed255570f57cc2a222962d55a90b79bd","project":"openstack/election","branch":"master","topic":"add-frickler-tc-candidacy","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-20 07:43:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5757771fed255570f57cc2a222962d55a90b79bd","subject":"Add Jens Harbott candidacy for TC 2025.1","status":"MERGED","created":"2024-08-19 20:25:01.000000000","updated":"2024-08-20 07:45:36.000000000","submitted":"2024-08-20 07:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926581-add-frickler-tc-candidacy","meta_rev_id":"4874354afc0664291bdaabee01aedb9f3e6cd514","_number":926581,"virtual_id_number":926581,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-20 07:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-08-19 21:00:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2024-08-20 07:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2024-08-20 07:32:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 20:34:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-19 21:00:51.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2024-08-20 07:32:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a516927a7479da84e544ad6b7884dae331e819b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-19 20:25:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb5c55c3f23a809589ef84f6efae4dbd291795ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 20:34:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a31c7370ed324f28a7be1970069b3a20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f4aac66fbe4d35a7eff3b041887894 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ae26b78c6d54d54899e71cce2ad8843 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37940a2d9eb54752969d681f4d760bdd : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d2bad34363c4dee844b1a949d6160cf : SUCCESS in 2m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/94a029e6d5964c97a00531b2e35229f6 : SUCCESS in 2m 40s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/7523cc4a6e514eb18f78b362157624ce : SUCCESS in 2m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b3086110891a4cccb06165b28e0e26b5 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"31f8b5028ca428d691c4190fe40de82e7def334c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-19 21:00:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6aaa14b46163dc68b0a3c7d18d91959681f1f4e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-20 07:32:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a39d6c2bb9e40d8012a387002bf1f9d1483272f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 07:33:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4fc1312bd501b986c898ad9cfc73b8538ef504d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 07:43:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c4e14a0abe1425c95001e4efa83dbd5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d41c95a38f84f07959aa50d5786737f : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ede95ebc4a8549ab9afa5de63cc86f2a : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b72d761cf1ff45a880ef7cb3e5629345 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91a1b8ff49d34f1bb6d50e108dbc48ad : SUCCESS in 4m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6982f2b358944a8a9faa0a6115c79fd8 : SUCCESS in 3m 41s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/422b677a8631493dae4526dc13f22f15 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b80d8090f9dbaacf9098522db10bb39e18213f57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 07:43:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4874354afc0664291bdaabee01aedb9f3e6cd514","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 07:45:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/849f8fbf8de649709512c0a3c1f4780e\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/5539d7c3abf54f669079c2235091041c : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e880ff06b9c1abf338f9451d1cd606da923ae8dd","revisions":{"e880ff06b9c1abf338f9451d1cd606da923ae8dd":{"kind":"REWORK","_number":1,"created":"2024-08-19 20:25:01.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/81/926581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/81/926581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/81/926581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/81/926581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/81/926581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/81/926581/1"}}},"commit":{"parents":[{"commit":"9535df097d1996bc898d0dcb6cc986b6104c228c","subject":"Merge \"[trivial] Remove \"import input\" on check-all-candidacies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9535df097d1996bc898d0dcb6cc986b6104c228c"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-19 20:24:05.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-19 20:24:05.000000000","tz":120},"subject":"Add Jens Harbott candidacy for TC 2025.1","message":"Add Jens Harbott candidacy for TC 2025.1\n\nChange-Id: I5757771fed255570f57cc2a222962d55a90b79bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e880ff06b9c1abf338f9451d1cd606da923ae8dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/e880ff06b9c1abf338f9451d1cd606da923ae8dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
