)]}'
{"id":"openstack%2Felection~927237","triplet_id":"openstack%2Felection~master~Ia05db5b47c54ed8883ddd301cf25227f2cc3a879","project":"openstack/election","branch":"master","topic":"ptl-candidacy","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-08-27 13:05:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia05db5b47c54ed8883ddd301cf25227f2cc3a879","subject":"Add Pranali Deore candidacy for Glance PTL 2025.1","status":"MERGED","created":"2024-08-27 07:43:04.000000000","updated":"2024-08-27 13:06:24.000000000","submitted":"2024-08-27 13:05:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927237-ptl-candidacy","meta_rev_id":"3eece73761c66fa7852eb5d5a4e87f6fce2e139a","_number":927237,"virtual_id_number":927237,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-27 13:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-08-27 12:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2024-08-27 09:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2024-08-27 12:42:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-27 07:51:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-27 09:52:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-08-27 12:42:41.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"6e3cb67ee473b3e0cd4b4b62445741c0013cbe03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-08-27 07:43:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a86976c7c5e43c793075f8b76c347eabb327b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 07:51:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7517776bf91486c9ec12c6ddf93f6e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d05f9d8e4641eaa6c2edee567e9d26 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0eaa7ae953274b90bc677878e6af2b31 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75c92fce266f4185b969c3febcf50692 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2de6360a61fa4abfb2b3607e2995592b : SUCCESS in 2m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6056a3ab713c474796b48851bd24cbef : SUCCESS in 2m 37s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/3902c5e4e03947b684c982fdc40ed0ce : SUCCESS in 2m 40s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/578e0f77c60647e498961ab69f4bc30e : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1799d508f142fd197ef0026c9976ed3a11d207a7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-27 09:52:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf72f8a047c4fa9da70afff347808e8198e842e5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-27 12:42:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2dbc0d6d13574974b48635e0a87cc47c8ee04f15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 12:43:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09c3e571d4c7e749f28c423480da218594f85e1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 13:05:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/942de844c4ab49e0a8ee844cc4ae19de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f3688a8b84443eb2d6f0ca7d1b3ccb : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb274e167ede4ad0a2054bde196e0424 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/735319ba35364b3593c084c88f3b906e : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3a2d13fd3524cba808ed90911c503bf : SUCCESS in 4m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7c2e01fb687a436a8c6be91b817875f3 : SUCCESS in 4m 25s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5e863045d3bd4d3e8bcb3a9e61749836 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"811c3a6d65b47e4bf3286298a9fcc80c308d0567","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 13:05:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3eece73761c66fa7852eb5d5a4e87f6fce2e139a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 13:06:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a17454204284d16a86c705150913085\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/14652dc5aa894be291b0981983b71c21 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcbe6619388c2352eda55916d8bff62318b6b27e","revisions":{"dcbe6619388c2352eda55916d8bff62318b6b27e":{"kind":"REWORK","_number":1,"created":"2024-08-27 07:43:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/37/927237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/37/927237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/37/927237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/37/927237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/37/927237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/37/927237/1"}}},"commit":{"parents":[{"commit":"657bfd5d2790e10b0107cbc04f10b3871a37d378","subject":"Merge \"Add Yasufumi Ogawa candidacy for Tacker PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/657bfd5d2790e10b0107cbc04f10b3871a37d378"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-08-27 07:39:17.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-08-27 07:39:17.000000000","tz":0},"subject":"Add Pranali Deore candidacy for Glance PTL 2025.1","message":"Add Pranali Deore candidacy for Glance PTL 2025.1\n\nChange-Id: Ia05db5b47c54ed8883ddd301cf25227f2cc3a879\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dcbe6619388c2352eda55916d8bff62318b6b27e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/dcbe6619388c2352eda55916d8bff62318b6b27e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
