)]}'
{"id":"openstack%2Felection~927280","triplet_id":"openstack%2Felection~master~Iafa8f26df27976dd5d916a03413fa602e72492a5","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-08-28 14:46:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2024-08-28 15:19:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iafa8f26df27976dd5d916a03413fa602e72492a5","subject":"Added Walt Boring candidacy for Cinder","status":"MERGED","created":"2024-08-27 14:36:02.000000000","updated":"2024-08-28 15:21:34.000000000","submitted":"2024-08-28 15:19:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927280","meta_rev_id":"ee53db5632c5f0f54cfc67dfde50eea06d325510","_number":927280,"virtual_id_number":927280,"owner":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 15:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-08-28 15:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2024-08-28 14:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2024-08-28 15:10:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-27 14:48:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-27 17:50:36.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-08-28 15:10:57.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"e61735e672a1279345ba8966fa094f14b4284c7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2024-08-27 14:36:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1cd9b583c36e6a9d24ba13994bcb696760f6f6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 14:48:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9430472cfa80425ca75345db557527d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30e5e767e1ff4ca8b7914b8700a18037 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/549f24069e474d059df2310e075aa6e1 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d33118c66afd4865922fad439adc6668 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd308b8867224d8594327a9658d79e1c : SUCCESS in 3m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f3e191400490473184334e59e41a20cb : SUCCESS in 2m 58s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/815da364088e4aab8ce1518e98a004a2 : FAILURE in 4m 09s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/59b2447b5e1f404d919f40dd336c7412 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"845a36f6f87bb07e96b686c3af61b4ac0a322721","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-27 17:50:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fff9c18bdeaee295f1584b4baa57bb8d1e2c1492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2024-08-28 13:16:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3bedc6d10322eb4eb2c66995c7b671c013d5c434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 13:40:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ff4a240b08e47f4a10f1fa56a36ee8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d456442f284dc88346c97a4ed07f3f : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c40967f740a04523b1a1074c1e9a71ed : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03e8a6ebc2574e8084dc50664422ac78 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6da537a235044d6b9e2160f460d8414 : SUCCESS in 3m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3dc25aa2214544c19517bed5a9cbf64b : SUCCESS in 3m 13s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/4171e68014024827836f96e022b74949 : SUCCESS in 3m 06s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/cfba4f3b26df4fe68910b03a727e8147 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ece5467cf53dc0fed9b84478bbfa2d750bf48753","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-28 14:46:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"05aff9b0b28754ad7cf30bb61f680e6e25e7ceb8","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-28 15:10:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57a8529d66c6ae465e5c34c9123bcc5d1ea0259e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:11:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7fda746ab7282d904633e2c248e7d8754687618a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:19:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8350a608bca8478c99635697a7d86148\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7ccec6e38c4d1ab9e01d42b844919b : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72bb629cb91648e4bba388a2d67f1ef2 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45c9ba80acf74c0aa26738918e416790 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6047bbf519e45a6a385251e3efa2f31 : SUCCESS in 3m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b3095da164f84a8eabba71c0e3322d7a : SUCCESS in 3m 56s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/b2343c2e8a294331a024b6c80fe2b662 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f1846e343899f1db60282f6e49155c17c08f8e8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:19:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ee53db5632c5f0f54cfc67dfde50eea06d325510","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:21:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca852266165b496b808329dc8a91ce1a\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/acd59460985342aaaaad8e6978024ce5 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d2d2b3a41f963221567561b6305575ad194efa9","revisions":{"17d7091589d22125b16e28a4501e162fc87989c4":{"kind":"REWORK","_number":1,"created":"2024-08-27 14:36:02.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/80/927280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/80/927280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/80/927280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/80/927280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/80/927280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/80/927280/1"}}},"commit":{"parents":[{"commit":"3f6637d2457ca14e23e649bf268975204396f26c","subject":"Merge \"Adding Takashi Kajinami candidacy for Heat PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3f6637d2457ca14e23e649bf268975204396f26c"}]}],"author":{"name":"Walter Boring IV","email":"walter.boring.iv@sap.com","date":"2024-08-27 14:33:17.000000000","tz":-240},"committer":{"name":"Walter Boring IV","email":"walter.boring.iv@sap.com","date":"2024-08-27 14:36:01.000000000","tz":-240},"subject":"Added Walt Boring candidacy for Cinder","message":"Added Walt Boring candidacy for Cinder\n\nThis patch adds Walt Boring as candidate for Cinder PTL\n\nChange-Id: Iafa8f26df27976dd5d916a03413fa602e72492a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/17d7091589d22125b16e28a4501e162fc87989c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/17d7091589d22125b16e28a4501e162fc87989c4"}]},"branch":"refs/heads/master"},"8d2d2b3a41f963221567561b6305575ad194efa9":{"kind":"REWORK","_number":2,"created":"2024-08-28 13:16:45.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/80/927280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/80/927280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/80/927280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/80/927280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/80/927280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/80/927280/2"}}},"commit":{"parents":[{"commit":"3f6637d2457ca14e23e649bf268975204396f26c","subject":"Merge \"Adding Takashi Kajinami candidacy for Heat PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/3f6637d2457ca14e23e649bf268975204396f26c"}]}],"author":{"name":"Walter Boring IV","email":"walter.boring.iv@sap.com","date":"2024-08-27 14:33:17.000000000","tz":-240},"committer":{"name":"Walter Boring IV","email":"walter.boring.iv@sap.com","date":"2024-08-28 13:16:33.000000000","tz":-240},"subject":"Added Walt Boring candidacy for Cinder","message":"Added Walt Boring candidacy for Cinder\n\nThis patch adds Walt Boring as candidate for Cinder PTL\n\nChange-Id: Iafa8f26df27976dd5d916a03413fa602e72492a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8d2d2b3a41f963221567561b6305575ad194efa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/8d2d2b3a41f963221567561b6305575ad194efa9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
