)]}'
{"id":"openstack%2Felection~927506","triplet_id":"openstack%2Felection~master~I1a378952bcdc308ae70a3a5a58c1ab61466d72a6","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2024-08-30 15:25:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a378952bcdc308ae70a3a5a58c1ab61466d72a6","subject":"Withdrawing Walt Boring Candidacy for cinder PTL","status":"MERGED","created":"2024-08-29 13:15:09.000000000","updated":"2024-08-30 18:28:46.000000000","submitted":"2024-08-30 15:25:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927506","meta_rev_id":"2e46b7115d26099171a009cc9e49e9c12a3adcf5","_number":927506,"virtual_id_number":927506,"owner":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 15:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-08-30 15:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2024-08-29 20:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2024-08-30 15:01:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-29 13:29:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-29 16:35:57.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"CC"},{"updated":"2024-08-29 20:46:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-08-30 15:01:04.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2024-08-30 18:28:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"ff503ecee4a273fb4dac88a3181d46eb749a5e13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2024-08-29 13:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b05b9f83bc495d0fc225adeae21d3703463610c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 13:29:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e9dbfe9ba5471ba283b8abca76476a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c92f09a4681f4f65bdbfe40f2051a1ef : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f631fbf819ef4fa886aeb91326b9dac8 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d8b711b2a62486b98f837f58e4072ed : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/677751702a70429a8eca059cf7718261 : SUCCESS in 3m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/01842e1f41a04939b9b5709a3757fd2d : SUCCESS in 4m 09s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/923ba33ed6834a0d840cc0b0a6eb7128 : SUCCESS in 2m 47s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/28ede2e3be374f87b354f3a94f313b26 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"af90f5cf15655474774283f45b2764ce40b33f11","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-08-29 16:35:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83d17b7f5df9e0d05da2d2ba7491f38be1888d74","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-29 20:46:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9476db8635f4eb7d32569464ae9d0404f3d75e3","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-30 15:01:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae2e079093b4af44fd461c3edf5cc7045fccfddb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:01:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"daf193b4c8d3afebffaa57bb32a98dc75b95f40e","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-08-30 15:03:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"395bbaef2b2feedcee1c3554f501f7b7a1018280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:25:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f79520e39054381b54e668d09305aae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76c2e6573414a1482d3081884c2f856 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a17c59d54d834233aab83ec7b8de16f0 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25aa2b894ba34955a52a3bb4913cc7f5 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01cfc43950154c2c98a17c9e1a9d3e72 : SUCCESS in 3m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e8f23e3335ae4deaae2747fb74aec2c9 : SUCCESS in 3m 29s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/d1c69967984445ca844643f8ceef328a : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"90a054ccbfdac36dcda93b08f654c10a7b4abc96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:25:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c564b570e8b08e3fbdd1232e2d09d9258db5b069","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:27:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe75a932ab1477eb309674b336daac5\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/8fa94010a56a417ba459f8e29ade8b6f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"2e46b7115d26099171a009cc9e49e9c12a3adcf5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 18:28:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e77196ddf40ac688fd292da84e3573977feef5b","revisions":{"1e77196ddf40ac688fd292da84e3573977feef5b":{"kind":"REWORK","_number":1,"created":"2024-08-29 13:15:09.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/06/927506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/06/927506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/06/927506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/06/927506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/06/927506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/06/927506/1"}}},"commit":{"parents":[{"commit":"b597cf779ef238bf8cf2af3f22eccded6b2b3ff6","subject":"Merge \"Add Jon Bernard candidacy for Cinder PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b597cf779ef238bf8cf2af3f22eccded6b2b3ff6"}]}],"author":{"name":"Walter Boring IV","email":"walter.boring.iv@sap.com","date":"2024-08-29 13:13:51.000000000","tz":-240},"committer":{"name":"Walter Boring IV","email":"walter.boring.iv@sap.com","date":"2024-08-29 13:13:51.000000000","tz":-240},"subject":"Withdrawing Walt Boring Candidacy for cinder PTL","message":"Withdrawing Walt Boring Candidacy for cinder PTL\n\nI am withdrawing my candidacy for cinder PTL as Jon Bernard\nhas put his candidacy in.  I just wanted someone on Cinder\nto be PTL.\n\nChange-Id: I1a378952bcdc308ae70a3a5a58c1ab61466d72a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1e77196ddf40ac688fd292da84e3573977feef5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/1e77196ddf40ac688fd292da84e3573977feef5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
