)]}'
{"id":"openstack%2Felection~940822","triplet_id":"openstack%2Felection~master~Ic92672d339a9ee3da9f4026c3b3363b4cb8aabb7","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-02-06 16:53:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic92672d339a9ee3da9f4026c3b3363b4cb8aabb7","subject":"Adding René Ribaud candidacy for Nova 2025.2 PTL","status":"MERGED","created":"2025-02-05 18:30:34.000000000","updated":"2025-02-06 18:27:23.000000000","submitted":"2025-02-06 16:53:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940822","meta_rev_id":"d35037b1404ed974ff996f52793cdf9564a35bd5","_number":940822,"virtual_id_number":940822,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-06 16:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-02-06 16:41:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2025-02-06 18:27:23.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-02-06 08:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-02-06 16:41:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-05 18:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-06 08:54:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-02-06 16:41:33.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-02-06 18:27:23.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"a4083fc3471a059da81047d8e2063afcbd19ab3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-02-05 18:30:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67a0cc0183592479b5d4ff883b65b5f28ab23a9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 18:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c263b1e4d85541a789c118fa8bb17ecf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b49c69c891641949c5928b7a9cfb639 : SUCCESS in 2m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffdacbdf6eb3475b8dd3367557134fd7 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2df60adb628e4e01ab47a45d84956ac4 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2baf84c1c7cb4067b541cb38c2946481 : SUCCESS in 2m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ebeea32a059e4fe4835a98e848898e82 : SUCCESS in 3m 54s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/093340e7080541c78b6d269d0f23a60b : SUCCESS in 3m 16s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/23de8100f42e4817b9a7fabe1510b5e1 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ebca6c7428d5005ba80562f7d5d9915b09ed8e25","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-06 08:54:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b313fe411858a38bb825938b8abb9968d132102","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-06 16:41:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74b61a514771dd44ce5f29604facae7ba19fc0d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 16:41:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97aeff77e153ecf4f5acbbdd0a9dd46d8112ff65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 16:53:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26c3bce7601b4728951e39b7a80d7c7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af62b93b8bd84dbb89ea44b43d45bc05 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ddbb20a555840d9a00279be20c043ac : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d71675ae21e446ddaea739fb52fa9d89 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c5a6340d01548a7a40d2e4afe743b2a : SUCCESS in 3m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7ae4d360dfa241caa236ff65ec49447c : SUCCESS in 3m 46s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ec94c3785b274d5cbea6919d6dd9e3eb : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a8894b608b09518c977ca933e21c159d012a83b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 16:53:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"022553cc5192a68ee9c443952636578441bf8a7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 16:54:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08f58b6ab22a43808f3ea8aeb0bb7a80\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/3e10da77f72a44e783da99f308927d95 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1},{"id":"d35037b1404ed974ff996f52793cdf9564a35bd5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-02-06 18:27:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c1222bfac44f03ffe45aa1a15cc3e9a12f13a46","revisions":{"2c1222bfac44f03ffe45aa1a15cc3e9a12f13a46":{"kind":"REWORK","_number":1,"created":"2025-02-05 18:30:34.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/22/940822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/22/940822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/22/940822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/22/940822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/22/940822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/22/940822/1"}}},"commit":{"parents":[{"commit":"ea473d485ddbc8ca768eaa5d96abdda9874c7134","subject":"Add 2025.2 election nomination dir for oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ea473d485ddbc8ca768eaa5d96abdda9874c7134"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-02-05 18:29:49.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-02-05 18:30:32.000000000","tz":60},"subject":"Adding René Ribaud candidacy for Nova 2025.2 PTL","message":"Adding René Ribaud candidacy for Nova 2025.2 PTL\n\nChange-Id: Ic92672d339a9ee3da9f4026c3b3363b4cb8aabb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2c1222bfac44f03ffe45aa1a15cc3e9a12f13a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2c1222bfac44f03ffe45aa1a15cc3e9a12f13a46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
