)]}'
{"id":"openstack%2Felection~941094","triplet_id":"openstack%2Felection~master~I09f9b2e64f3cf0c13ae06e212fef88f856f7fb3d","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-02-10 22:11:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09f9b2e64f3cf0c13ae06e212fef88f856f7fb3d","subject":"Add Vladimir Kozhukalov candidacy for Openstack-Helm 2025.2 PTL","status":"MERGED","created":"2025-02-10 12:45:45.000000000","updated":"2025-02-10 22:12:33.000000000","submitted":"2025-02-10 22:11:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941094","meta_rev_id":"2f1fd92cd1e516dd6027c5ceba58e17f046b184c","_number":941094,"virtual_id_number":941094,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-10 22:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-02-10 21:54:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-02-10 14:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-02-10 21:54:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-10 12:54:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-10 14:25:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-02-10 21:54:03.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"da514517926c4deb1545bbc75225bb3e683dd900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-02-10 12:45:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c343fab22e2e352fa5a7c6e0f5b8b06a84066ead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-02-10 12:46:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"51bd40184a750c32291133b896d10cfbbc0bf4a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 12:54:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f86cd9ccca433ea61d25e382a55700\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d68be12cc14e90afc21e31a353574b : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae47bb895ba74cd8b2ffd9ab7424d122 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4382b7b3f4bf4595a8d68885c33011c9 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d9c7935e6f64ebaad7db734356cc2f3 : SUCCESS in 3m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/50f4bb73427741f3ad1abfb77467cd94 : SUCCESS in 3m 12s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/9a8af0a18c3748b4ad81da0a0d1ace2c : SUCCESS in 4m 33s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/40d9527cc3ef4203be69d02ff457da44 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d75d4a0364f3ab0c205f89366841a4776cb01b06","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-10 14:25:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"726457d167e2067a0a669ae5353fee0942d900f8","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-10 21:54:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb423534a04079675713410c818b33025ba2b2ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 21:54:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8bb36dce060d4bd8d4ac4e52203ea6b69a979719","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 22:11:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/758a7aefa709427d92ee4ccab14c4325\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a91398794e4b0b9f4c8453e3c9c5bb : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88eadfbe6204474f9fe81143d04a0804 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8819b94511ae4630a46cca2985c68b63 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54f00c49683c4b19b412381e60bf46db : SUCCESS in 2m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/132105d52b5a452e8115f585df473759 : SUCCESS in 2m 52s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/edf2fa49cc4448a49408e92da26f4a89 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"eebec419ce7edc290b7637bbbb9448ad049d76c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 22:11:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2f1fd92cd1e516dd6027c5ceba58e17f046b184c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 22:12:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fe164be04904f0390955720433ab262\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/0293e472408349f69b8a4e00e5eb7f63 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee549c0f6bda8fd67ef8cfc162346b37d7a459aa","revisions":{"ff827b60e7012bbb426c76e2dab9b99ff43e5b0a":{"kind":"REWORK","_number":1,"created":"2025-02-10 12:45:45.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/941094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/94/941094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/94/941094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/94/941094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/94/941094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/94/941094/1"}}},"commit":{"parents":[{"commit":"b979728f1b49b1a2f29fd2eb826d15edcac2e57b","subject":"Merge \"Adding Omer Schwartz for Designate 2025.2 PTL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b979728f1b49b1a2f29fd2eb826d15edcac2e57b"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-02-10 12:45:23.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-02-10 12:45:23.000000000","tz":-360},"subject":"Add Vladimir Kozhukalov candidacy for Openstack-Helm 2025.2 PTL","message":"Add Vladimir Kozhukalov candidacy for Openstack-Helm 2025.2 PTL\n\nChange-Id: I09f9b2e64f3cf0c13ae06e212fef88f856f7fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ff827b60e7012bbb426c76e2dab9b99ff43e5b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ff827b60e7012bbb426c76e2dab9b99ff43e5b0a"}]},"branch":"refs/heads/master"},"ee549c0f6bda8fd67ef8cfc162346b37d7a459aa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-10 12:46:03.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/94/941094/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/94/941094/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/94/941094/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/94/941094/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/94/941094/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/94/941094/2"}}},"commit":{"parents":[{"commit":"67a947a22174e3556e998ca6b1b6051e91b01c71","subject":"Merge \"Adding Dale Smith candidacy for Adjutant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/67a947a22174e3556e998ca6b1b6051e91b01c71"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-02-10 12:45:23.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-02-10 12:45:53.000000000","tz":-360},"subject":"Add Vladimir Kozhukalov candidacy for Openstack-Helm 2025.2 PTL","message":"Add Vladimir Kozhukalov candidacy for Openstack-Helm 2025.2 PTL\n\nChange-Id: I09f9b2e64f3cf0c13ae06e212fef88f856f7fb3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ee549c0f6bda8fd67ef8cfc162346b37d7a459aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ee549c0f6bda8fd67ef8cfc162346b37d7a459aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
