)]}'
{"id":"openstack%2Felection~941175","triplet_id":"openstack%2Felection~master~Iadfbc0ce536e48dba876d7c21cb537b6491fa4b7","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2025-02-12 01:39:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iadfbc0ce536e48dba876d7c21cb537b6491fa4b7","subject":"Adding Eric Zhang candidacy for Venus","status":"ABANDONED","created":"2025-02-11 00:55:20.000000000","updated":"2025-02-12 01:39:35.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0ab07921fdfdeb2dcfe6b52303a299c4131919cd","_number":941175,"virtual_id_number":941175,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-11 01:07:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":-1,"date":"2025-02-11 13:36:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 01:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-11 13:36:24.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"415fc94252b4b4638ea000bc074950ba0ff8909b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2025-02-11 00:55:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d7de73785a742edfd62c854af91556236ad3fea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 01:07:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42eb65b89e394f9e9997a3ad31ee26d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412f3ba63b7f4b8884c4323e0c5fa96d : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1906c796b5454915801f6ad8503e1e6f : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7d2eb4b2e984565b42484d6bf28b774 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d4aadbe95ba4630ae0c803ed9df2881 : SUCCESS in 3m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8708d7a400574ce6ae1c806d6d3a0cf8 : SUCCESS in 3m 10s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/1727fcb8e3cb47558d02c02ab636cc15 : FAILURE in 3m 16s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/fa98dbbe7b034a24ad5cc7f895d62ac8 : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"13dcc35e8a8f6940bd9ec123c78c4a97b614d245","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-11 13:36:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3896926e62a434b53a8a486cc2a0f0c3bec4df5","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2025-02-12 00:35:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95327e6d2753ab7783fd6e428b46db75c82187cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 00:59:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b875204366a4755b2d4570a5b737b18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9c09ffa6f14e0d983cd781f6230c53 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cd2c02f170d42f38c89723b6f0cb14a : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a3c14639d3a4def9b2f4ab7ef40e950 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94e02f8cd655402da051cadd74801649 : SUCCESS in 3m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0851109014aa491d847cb0e470c8d3ae : SUCCESS in 4m 34s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/a94c111f58e448dca082b4f3fc26ecf0 : FAILURE in 2m 17s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/3127d95e5891458ca8a4fb8e7b8ac3e6 : FAILURE in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"34c6071ae70bc064ae4b31a4ccc9c9c1080b939f","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2025-02-12 01:00:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b2804dca34ccd382cd47bd3c38537aa2e13db55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 01:27:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5373d5fad1e74191ab5ede4afd163d35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb380f2fae0441288e4e469c51397bd : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5350a77924f54d7e8d0d3e56d0ec585b : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3a6d93180234783a592e5dabd79bcb9 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/770cc477c4e949febcb0eb171795726b : SUCCESS in 4m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e83ce9da6c3640aaa2b819155d85267d : SUCCESS in 3m 21s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/f3788b292ab6433285bff9eacc1c75b8 : FAILURE in 4m 26s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/c3917d2d866148db9db2068e9733a4bf : FAILURE in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c8030a4ed51c1d1a3a6dc04776f38612fa97415b","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2025-02-12 01:28:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"626464bbd4e196f0cadde866df88236eb9d7d11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 01:31:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4d92acc19784407971d55a08ab45150\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4deb51013664837af74e570939b1aac : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0c05499950a47ffa6b230a3595aab8c : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b68920b951d34fb6aec1dceeed098796 : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05b7d93b07d04869a2dc57d34f23314b : SUCCESS in 2m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d936e186714d4e01b43d4dfe377d1866 : SUCCESS in 2m 19s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/6902e23781fb4c529db9d97f4d821746 : FAILURE in 2m 21s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/5a6740c27c46463db7d7a308eeeb28d5 : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0ab07921fdfdeb2dcfe6b52303a299c4131919cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2025-02-12 01:39:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96d37a801dc827caa6ba237340a32ad20987f13b","revisions":{"96d37a801dc827caa6ba237340a32ad20987f13b":{"kind":"REWORK","_number":1,"created":"2025-02-11 00:55:20.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/75/941175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/75/941175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/75/941175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/75/941175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/75/941175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/75/941175/1"}}},"commit":{"parents":[{"commit":"ee549c0f6bda8fd67ef8cfc162346b37d7a459aa","subject":"Add Vladimir Kozhukalov candidacy for Openstack-Helm 2025.2 PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ee549c0f6bda8fd67ef8cfc162346b37d7a459aa"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2025-02-11 00:55:01.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2025-02-11 00:55:01.000000000","tz":0},"subject":"Adding Eric Zhang candidacy for Venus","message":"Adding Eric Zhang candidacy for Venus\n\nChange-Id: Iadfbc0ce536e48dba876d7c21cb537b6491fa4b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/96d37a801dc827caa6ba237340a32ad20987f13b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/96d37a801dc827caa6ba237340a32ad20987f13b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
