)]}'
{"id":"openstack%2Felection~941612","triplet_id":"openstack%2Felection~master~I20b14f4e2bbd805c9ebe530cbd3cded7f2cf1574","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2025-03-11 14:12:29.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-05-12 07:32:43.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-05-12 07:26:16.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"I20b14f4e2bbd805c9ebe530cbd3cded7f2cf1574","subject":"Add tc repositories for electoral roll generation","status":"MERGED","created":"2025-02-13 23:17:07.000000000","updated":"2025-05-12 07:33:25.000000000","submitted":"2025-05-12 07:32:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941612","meta_rev_id":"d6471506f6dddb125f940a5ae7a8431cdb56cd1e","_number":941612,"virtual_id_number":941612,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-12 07:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-03-11 14:12:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-05-12 07:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-11 14:34:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2025-05-12 07:26:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 23:24:55.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-02-13 23:28:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-14 08:57:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-02-14 22:21:16.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-03-11 14:34:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"b778c659bf1581109af2f78b809e5931a20c0f83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-02-13 23:17:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b4058aa418cc201dea0c714e7af500c5d616463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-02-13 23:19:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6d7efc885dfcd20286cb523e8aabd4af2301d3b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-13 23:24:55.000000000","message":"Patch Set 2:\n\nSkimming over this, it might help if the technical-committee-repos.yaml file in governance included a placeholder or something for extra ACs, since I\u0027m concerned that if some ever do get added to that file they could end up just about anywhere in the schema (at best silently getting missed when building the electorate, at worst breaking the election scripts). It would not be the first time that edits to structured data in governance broke these tools.","accounts_in_message":[],"_revision_number":2},{"id":"01f1dc2fa171b7f3db3ae5dbc876e01b7abc6ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 23:28:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/313dc041d4d04023924ee2c5e881dd92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5516307ae1493c9b73e6ca8f217a84 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cea1169ebd2a4a0ea62ef2b211acd3f6 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d220e808b5d54a1abf17793789f1a9a5 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d07f0a3fa014864bdcda8cf881e7b0a : SUCCESS in 3m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/804312f986a2489c85a941178695fea2 : SUCCESS in 2m 34s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/dd49e0be6189470d8a9aafab21e09c2d : SUCCESS in 2m 22s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/af60c598381d4a3e91d30a27fdec8728 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a3c984259d67da561d2dd5e8f01accc969cbb811","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-14 08:57:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ffaef2de7da62211bc0ee8510a9e84c1d3eda138","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-14 22:21:16.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fce1a87b564e3636b695b2be8eed65db62fac6af","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-04 18:08:44.000000000","message":"Patch Set 2:\n\nI\u0027ll plan to use this when I generate the list of 2025.1/Epoxy contributors at the end of next week, but if it merges before then that will help me to not forget.","accounts_in_message":[],"_revision_number":2},{"id":"0dd0d1626a61e0de79626ad75e2c86dea87ec2da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-04 20:34:35.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f294ea9133993dd1a5d15e3877f42fe419ad8fd7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-04 20:37:33.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a44c137d6097094425db99338990cf6c6198896d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 20:46:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ccd1b23ef294e52ae6e5a66b9b4da0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6f924efd174dac908dea997a52c4c9 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b81f3b4e9f7a4b75982d71a78ab6402e : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b14cfb45d207438a9ed10106b5363f1f : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb4e3622aaea493fac48afe93e20cb59 : SUCCESS in 4m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b1d9f91e7cd147ee9e139af7aac87220 : SUCCESS in 3m 05s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/12add3d7de634c77aa7e84233382366e : SUCCESS in 4m 44s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/1456eaec7463440c862aea388824a186 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":3},{"id":"fd112f747e954abb00952736ecfe7456a4244e08","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-03-11 14:12:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7e3240afda8f267b4f32a529a3e12049b5893f72","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-03-11 14:34:56.000000000","message":"Patch Set 3: Code-Review+1\n\nI used this when I generated the contributors list for the upcoming 2025.1/Epoxy release page, and it seemed to work fine.","accounts_in_message":[],"_revision_number":3},{"id":"e459cf9f658b549bd2e53e36aa3611df1a04092c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-18 18:44:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4697a6101478d8a78611b111f94b57acaf2bb0c3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-12 07:26:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"43d6b3b2309faebeeffe888906a5ea557919b366","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 07:26:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"99ed04bc909e9e547c6fb0b841cc1519b1d9ff15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 07:32:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/870135ba4a91460892fb7b33905bcdef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4d152d850dc4cafa80c5795141f5e06 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b121dc06d4ab48788b05b2376b96f83b : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd9d30b5c1be4d17811094ceb17b2094 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb396f66511e47ec8f2ee6fba03c09a1 : SUCCESS in 4m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c212e2ab69674cbda2960e954c07cbcc : SUCCESS in 4m 33s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/ad4072a921f340128d1ccd4389470938 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3de8d853e33ed327947f26ac8d6165ee37a69e36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 07:32:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d6471506f6dddb125f940a5ae7a8431cdb56cd1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 07:33:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/782e52facafa4c75a3d270a0805792c4\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/bb65eb0382244b44a3a6feb0961ef821 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b34651b84532dff947b411a31fb51aba2b5fae3f","revisions":{"9ba030110b6772cce127e8922cae06b0797735fe":{"kind":"REWORK","_number":1,"created":"2025-02-13 23:17:07.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/12/941612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/12/941612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/12/941612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/12/941612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/12/941612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/12/941612/1"}}},"commit":{"parents":[{"commit":"ee549c0f6bda8fd67ef8cfc162346b37d7a459aa","subject":"Add Vladimir Kozhukalov candidacy for Openstack-Helm 2025.2 PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ee549c0f6bda8fd67ef8cfc162346b37d7a459aa"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-02-13 23:16:59.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-02-13 23:16:59.000000000","tz":-480},"subject":"Add tc repositories for electoral roll generation","message":"Add tc repositories for electoral roll generation\n\nThe OpenStack TC Charter [0] states that \"Active Contributors\"\nare those that committed a change to a repository under the\ngovernance of the OpenStack Technical Committee. These repos\ninclude OpenStack projects [1] under various project teams,\nrepositories maintained by SIGs [2] and repositories maintained\nby the Technical Committee [3]\n\nAdd TC repos to the roll generation scripts.\n\n[0] https://governance.openstack.org/tc/reference/charter.html#voters-for-tc-seats-ac\n[1] https://governance.openstack.org/tc/reference/projects/index.html#projects\n[2] https://governance.openstack.org/tc/reference/technical-committee-repos.html#tc-repos\n[3] https://governance.openstack.org/tc/reference/sig-repos.html#sig-repos\n\nChange-Id: I20b14f4e2bbd805c9ebe530cbd3cded7f2cf1574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9ba030110b6772cce127e8922cae06b0797735fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/9ba030110b6772cce127e8922cae06b0797735fe"}]},"branch":"refs/heads/master"},"c0f5b3425a3af947e3b8bcf025bf6b3b755771b6":{"kind":"REWORK","_number":2,"created":"2025-02-13 23:19:30.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/12/941612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/12/941612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/12/941612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/12/941612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/12/941612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/12/941612/2"}}},"commit":{"parents":[{"commit":"ee549c0f6bda8fd67ef8cfc162346b37d7a459aa","subject":"Add Vladimir Kozhukalov candidacy for Openstack-Helm 2025.2 PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ee549c0f6bda8fd67ef8cfc162346b37d7a459aa"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-02-13 23:16:59.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-02-13 23:19:02.000000000","tz":-480},"subject":"Add tc repositories for electoral roll generation","message":"Add tc repositories for electoral roll generation\n\nThe OpenStack TC Charter [0] states that \"Active Contributors\"\nare those that committed a change to a repository under the\ngovernance of the OpenStack Technical Committee. These repos\ninclude OpenStack projects [1] under various project teams,\nrepositories maintained by SIGs [2] and repositories maintained\nby the Technical Committee [3]\n\nAdd TC repos to the roll generation scripts.\n\n[0] https://governance.openstack.org/tc/reference/charter.html#voters-for-tc-seats-ac\n[1] https://governance.openstack.org/tc/reference/projects/index.html#projects\n[2] https://governance.openstack.org/tc/reference/technical-committee-repos.html#tc-repos\n[3] https://governance.openstack.org/tc/reference/sig-repos.html#sig-repos\n\nChange-Id: I20b14f4e2bbd805c9ebe530cbd3cded7f2cf1574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c0f5b3425a3af947e3b8bcf025bf6b3b755771b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c0f5b3425a3af947e3b8bcf025bf6b3b755771b6"}]},"branch":"refs/heads/master"},"b34651b84532dff947b411a31fb51aba2b5fae3f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-03-04 20:34:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/12/941612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/12/941612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/12/941612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/12/941612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/12/941612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/12/941612/3"}}},"commit":{"parents":[{"commit":"ee549c0f6bda8fd67ef8cfc162346b37d7a459aa","subject":"Add Vladimir Kozhukalov candidacy for Openstack-Helm 2025.2 PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/ee549c0f6bda8fd67ef8cfc162346b37d7a459aa"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-02-13 23:16:59.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-03-04 20:34:34.000000000","tz":0},"subject":"Add tc repositories for electoral roll generation","message":"Add tc repositories for electoral roll generation\n\nThe OpenStack TC Charter [0] states that \"Active Contributors\"\nare those that committed a change to a repository under the\ngovernance of the OpenStack Technical Committee. These repos\ninclude OpenStack projects [1] under various project teams,\nrepositories maintained by SIGs [2] and repositories maintained\nby the Technical Committee [3]\n\nAdd TC repos to the roll generation scripts.\n\n[0] https://governance.openstack.org/tc/reference/charter.html#voters-for-tc-seats-ac\n[1] https://governance.openstack.org/tc/reference/projects/index.html#projects\n[2] https://governance.openstack.org/tc/reference/sig-repos.html#sig-repos\n[3] https://governance.openstack.org/tc/reference/technical-committee-repos.html#tc-repos\n\nChange-Id: I20b14f4e2bbd805c9ebe530cbd3cded7f2cf1574\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b34651b84532dff947b411a31fb51aba2b5fae3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/b34651b84532dff947b411a31fb51aba2b5fae3f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
