)]}'
{"id":"openstack%2Felection~941844","triplet_id":"openstack%2Felection~master~Ie5597f38c2851a0fda45d7952abcb390536ba464","project":"openstack/election","branch":"master","attention_set":{},"removed_from_attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2025-02-17 22:11:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5597f38c2851a0fda45d7952abcb390536ba464","subject":"Add Jon Bernard candidacy for Cinder PTL","status":"MERGED","created":"2025-02-16 19:53:52.000000000","updated":"2025-02-17 22:12:44.000000000","submitted":"2025-02-17 22:11:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941844","meta_rev_id":"4f834b6bd9a41b4b9ba729888d1b7f1f2b48a0ac","_number":941844,"virtual_id_number":941844,"owner":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-17 22:11:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-02-17 21:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-02-17 08:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2025-02-17 21:51:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-16 19:57:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-17 08:20:56.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-02-17 21:51:46.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"8d3d14e5bd4017c6a9541f639e2a39c55717d667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-02-16 19:53:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4901081a16655e11f505f1d756d722e48ae7a41d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-16 19:57:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52b97ea89dcb4adaae46de6eb69cc6e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/543e8743d9334e1ab9d4203fcc6bd314 : SUCCESS in 2m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/191e020188ae4c24a285447dc95562d6 : SUCCESS in 2m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65692881b5bb4d2baf627f7d405e4ba4 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4ed9f1fce1447a498e66c53d890b6cf : SUCCESS in 2m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/711066818572455785499cf8ec77d609 : SUCCESS in 2m 12s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/8b029eb9f83c477faead986965db6e46 : SUCCESS in 2m 58s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/06c3fd90c0894be9821c90c24f973d69 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c6bb224fe73c9c8e5ccd9a2773363917b6213fea","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-17 08:20:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9c3722cfbc5de5cde5e14950cb9b492d34e50bc","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-17 21:51:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9adbfe7afea126c72957322e424cd7e9f762f48a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 21:52:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec7151c6c79df886864ca9658d4099e03e8a4f16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 22:11:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f70aaae894f4cabbfbafc68e0f48ddb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88e8141abcfe4955b425fa7bb878d13a : SUCCESS in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf437f3d20bf4c8fa8024bed43d0c7fc : SUCCESS in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c43e1896f2454a1aaff3656676cceb46 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5092dc956c71489c8c00d66cafcd18c2 : SUCCESS in 2m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/221c4958eda240f9a5924be9f265fa4b : SUCCESS in 3m 20s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/e6badd0eddfd419096f9b06aa8ef18e8 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f99568b76e8884466a305b70df16e8d672afe7ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 22:11:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f834b6bd9a41b4b9ba729888d1b7f1f2b48a0ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 22:12:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0487b5710c574066bc12503b0a345879\n\n- promote-governance-election https://zuul.opendev.org/t/openstack/build/9b330e8362434591b6e1978079f7f48f : SUCCESS in 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ed511b53310d0271912718e5361e5fc359df659","revisions":{"2ed511b53310d0271912718e5361e5fc359df659":{"kind":"REWORK","_number":1,"created":"2025-02-16 19:53:52.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/44/941844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/44/941844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/44/941844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/44/941844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/44/941844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/44/941844/1"}}},"commit":{"parents":[{"commit":"c23d9b8da3f03316ab5cf088d87f7c96cd87a04d","subject":"spotz candidacy for TC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/c23d9b8da3f03316ab5cf088d87f7c96cd87a04d"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2025-02-16 19:53:25.000000000","tz":-300},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2025-02-16 19:53:52.000000000","tz":-300},"subject":"Add Jon Bernard candidacy for Cinder PTL","message":"Add Jon Bernard candidacy for Cinder PTL\n\nChange-Id: Ie5597f38c2851a0fda45d7952abcb390536ba464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2ed511b53310d0271912718e5361e5fc359df659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/2ed511b53310d0271912718e5361e5fc359df659"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
