)]}'
{"id":"openstack%2Felection~941937","triplet_id":"openstack%2Felection~master~I016c19eb8537fc18bada175145b54ccf2576e7c2","project":"openstack/election","branch":"master","topic":"ptl-candidacy","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-02-18 09:16:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I016c19eb8537fc18bada175145b54ccf2576e7c2","subject":"Add Cyril Roelandt candidacy for Glance PTL 2025.2","status":"MERGED","created":"2025-02-17 15:49:56.000000000","updated":"2025-02-18 09:16:48.000000000","submitted":"2025-02-18 09:16:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941937-ptl-candidacy","meta_rev_id":"52958c4763de04d70f552edf855f384da30b1557","_number":941937,"virtual_id_number":941937,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-18 09:16:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2025-02-17 21:56:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2025-02-18 08:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2025-02-18 08:24:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-17 16:17:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-17 21:56:01.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-02-18 08:24:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"3a13cbbeb1f34db6a7fbdd5fb4e1aa6d4cccaf15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-17 15:49:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2412c4e1e1faae09dc24d69dd6ce8e23a7f7aa56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 16:17:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ac6fbb3ec0d4ccdb2162b6930dcce5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd82aed38fd14e7b859a602b378854d4 : FAILURE in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45c359cf08d64f429359d1bcd67e6a5a : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8abb0fd16df49799efcfdf5ea3c5e9e : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6bdbf2107d34419bcac57010dd9728f : SUCCESS in 3m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0cc04f53efc3444c8d8f40c7d72d2f59 : SUCCESS in 3m 29s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/816ba39359904854a70ff9a024f0d0b1 : FAILURE in 4m 36s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/fb212771ce0143dbb8068215e1d3e393 : FAILURE in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1338833770b162874bd856036952f4957e089462","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-02-17 16:56:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11600b170dd35264e13e7838beb63ef6256fbc48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 18:00:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1925eae7e9384801bc519b9b7809c054\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dcf2297cb9f4cd09ab34d9945b16069 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86f08eecd87b4607842e319dd52ceeb8 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6492465864a48698de707a28be69aff : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6211e75f05e44856a3994a7d5bcfcb1d : SUCCESS in 3m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c2d3b1662cc43fca0da5f5fa47588cd : SUCCESS in 4m 12s\n- election-tox-ci-checks-review https://zuul.opendev.org/t/openstack/build/b0d8572d669b45418dd6319dc738fb56 : SUCCESS in 4m 19s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/a20a888896b041399e35f59353a29a22 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b558150be8a5635783ca98e3bfc3f0fd67642b44","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-02-17 21:56:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ab7140f742acee712b2a19fff5bc740741762c7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-02-18 08:24:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f2a84f2852648df46b8fdc53d9ccfcadb49e490","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 08:25:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"feeb5a453be086e47469f6dc81261901f860b6ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 09:16:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72585f405d8944af8aeff826972ef541\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16347bae36404978be2e5c1a4650dec7 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/134d88369b264f7784a17ddb25a9607a : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ed739e2092141c69f9454ccd1aa8125 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0915366d8b31436dbe24518d857524b8 : SUCCESS in 2m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/64ff28a6cc9c41018b8ec4d684edeae2 : SUCCESS in 3m 34s\n- election-tox-ci-checks-election https://zuul.opendev.org/t/openstack/build/9bb2d724417e4a2d9287bee0107a5848 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"52958c4763de04d70f552edf855f384da30b1557","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 09:16:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96bf42cd6dd809eddb939a4fea6adf9c95a52887","revisions":{"96bf42cd6dd809eddb939a4fea6adf9c95a52887":{"kind":"REWORK","_number":1,"created":"2025-02-17 15:49:56.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/37/941937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/election","ref":"refs/changes/37/941937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/election refs/changes/37/941937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/election refs/changes/37/941937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/election refs/changes/37/941937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/election refs/changes/37/941937/1"}}},"commit":{"parents":[{"commit":"305248d35aef892ac827b83ad03d3ad11dafa5e6","subject":"Add Brian Haley (haleyb) candidacy for Neutron PTL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/305248d35aef892ac827b83ad03d3ad11dafa5e6"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-17 15:47:51.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-02-17 15:49:54.000000000","tz":60},"subject":"Add Cyril Roelandt candidacy for Glance PTL 2025.2","message":"Add Cyril Roelandt candidacy for Glance PTL 2025.2\n\nChange-Id: I016c19eb8537fc18bada175145b54ccf2576e7c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/96bf42cd6dd809eddb939a4fea6adf9c95a52887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/election/commit/96bf42cd6dd809eddb939a4fea6adf9c95a52887"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
